)]}'
{"id":"openstack%2Foslo.limit~700338","triplet_id":"openstack%2Foslo.limit~master~I197d6486a3b126f616ffe2f7e3e669d922a1e337","project":"openstack/oslo.limit","branch":"master","hashtags":[],"change_id":"I197d6486a3b126f616ffe2f7e3e669d922a1e337","subject":"Trivial cleanup for tox","status":"MERGED","created":"2019-12-22 15:27:11.000000000","updated":"2019-12-23 16:15:56.000000000","submitted":"2019-12-23 16:14:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700338-1577117683359-a981c286","meta_rev_id":"6323e2d71ffe74bcfbf51fed80c04c1af196f62a","_number":700338,"virtual_id_number":700338,"owner":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-12-23 13:04:45.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-23 16:14:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":2,"date":"2019-12-23 13:27:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":1,"date":"2019-12-23 13:27:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-23 10:39:28.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2019-12-23 13:27:28.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-12-23 16:14:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"669cc7b012ac599f0b0c1c9fa072d80a83b42922","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-12-22 15:27:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea773a0e40913c35b4e42fbc9db60b002c05a66f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-22 17:09:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/ad3260d256454595a655f703be1ccde3 : SUCCESS in 1h 36m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0fcd51e8a8d74761a1c32fbe4d61cd1f : SUCCESS in 1h 16m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd3a6eb32f7c40f19648e2c1bf6fd5e8 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5053bc6db99f45419f607d8aae4a37c0 : SUCCESS in 2m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d147c171186945238c0b39a8d38343c3 : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66a7c664247f48b184a05fd7ba41bb3e : SUCCESS in 3m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4262aa02fa584411b6b51ab6a3aa96a2 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/784fe4cdff104e1e8fbcedb2be054fdd : SUCCESS in 3m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d941948e103848cf9be2d1dfa11dbcd7 : SUCCESS in 3m 04s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/b59e2eda70e842e3867bcd6304145dc3 : SUCCESS in 28m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f997951c33867db06fe5b210971585870e10bbeb","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-12-23 10:39:28.000000000","message":"Patch Set 1: Code-Review-1\n\nThese changes are being made to achieve community goal[1]. The commit message should include the details about it. Also the py27 drop for oslo libraries will start in phase-2 which should be soon.\n\n[1] https://etherpad.openstack.org/p/drop-python2-support","accounts_in_message":[],"_revision_number":1},{"id":"467974d90086e14ae5a9fa5735b1b08837737dc0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-23 12:35:55.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"532801984e0005b87963a07c4b78db38a993d0b3","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-12-23 13:04:11.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"fac66bfc5eebe1f0d0185ebb22e3979afa8b4052","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-12-23 13:04:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"97fdd410ad644dea1f70146da9effb4573e758ad","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-23 13:27:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2730e4cf60e24f2e5421a9e12221091da3d3dae9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 14:39:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/b1a3bb70317f4bb290a4bab791c148fc : SUCCESS in 1h 33m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5d63538a7084483a1ad91e5de4479d4 : SUCCESS in 1h 20m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48ac7b075df343a988cc064ee801292d : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5f2d6616e03465590701c6a86789708 : SUCCESS in 3m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b804b7d186494d7aa580e6baf949a8dc : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbf59be6332042fe8141b626e368097d : SUCCESS in 3m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a1d17277cc9349e9aed06cae37c9393a : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6367b4f13424887b954705a1be3b68a : SUCCESS in 3m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac594191c11c4d6c8a518bd2d269bf58 : SUCCESS in 3m 11s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/98cd6b02be9c4553be30b43689bbc035 : SUCCESS in 28m 23s","accounts_in_message":[],"_revision_number":2},{"id":"42313b5e8ef917894c30d05c83a4dc2e93797acb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 14:39:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"726749dc347c1d89e4ba30ab0c73c92687f28ca3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 16:14:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d6b914b1602542abbda49c1e54f4f5b6 : SUCCESS in 1h 34m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c458805bab084beaaa205af48a6cb01c : SUCCESS in 1h 22m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3e2a1088c08485ea800e503260e3010 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ac2820e89824e27b57f1c4e2a1f9620 : SUCCESS in 2m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3e8d91546db74eacac409222b1a021ee : SUCCESS in 2m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a619722cc36b4bbbb9b079bcc74dbd53 : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a61ddaa9a955439aa2c98db4eee8d64d : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ec8720e13454267a1c008e111ad0a4e : SUCCESS in 3m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/355279cf56ee4546bb2d712ea745c2eb : SUCCESS in 3m 04s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/97fdd1a9be1f48a4aca8d543d419f984 : SUCCESS in 30m 17s","accounts_in_message":[],"_revision_number":2},{"id":"7c8d77cbd7813aed582d3c0d174b0390e99c4113","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 16:14:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6323e2d71ffe74bcfbf51fed80c04c1af196f62a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 16:15:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/275b96975fbd464cb7e29a0eb2951d5e : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64ef84da599d4cc28410e16d3986d4fc : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"09eadf955e50eb6a3797a0a8c32c56d5622b3835","revisions":{"34d8dfb55fbc7f00ecbe5eb1c78855e3142159fc":{"kind":"REWORK","_number":1,"created":"2019-12-22 15:27:11.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/38/700338/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/38/700338/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/38/700338/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/38/700338/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/38/700338/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/38/700338/1"}}},"commit":{"parents":[{"commit":"51431f19fc5a42b28accb46a3402b861a7c00d2f","subject":"Merge \"Add flat enforcer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/51431f19fc5a42b28accb46a3402b861a7c00d2f"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-12-22 15:26:46.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-12-22 15:26:46.000000000","tz":480},"subject":"Trivial cleanup for tox","message":"Trivial cleanup for tox\n\nmove \u0027basepython\u0027 to the top-level \u0027testenv\u0027\n\nChange-Id: I197d6486a3b126f616ffe2f7e3e669d922a1e337\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/34d8dfb55fbc7f00ecbe5eb1c78855e3142159fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/34d8dfb55fbc7f00ecbe5eb1c78855e3142159fc"}]},"branch":"refs/heads/master"},"09eadf955e50eb6a3797a0a8c32c56d5622b3835":{"kind":"REWORK","_number":2,"created":"2019-12-23 13:04:11.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/38/700338/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/38/700338/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/38/700338/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/38/700338/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/38/700338/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/38/700338/2"}}},"commit":{"parents":[{"commit":"51431f19fc5a42b28accb46a3402b861a7c00d2f","subject":"Merge \"Add flat enforcer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/51431f19fc5a42b28accb46a3402b861a7c00d2f"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-12-22 15:26:46.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-12-23 13:04:09.000000000","tz":0},"subject":"Trivial cleanup for tox","message":"Trivial cleanup for tox\n\nmove \u0027basepython\u0027 to the top-level \u0027testenv\u0027\n\nChange-Id: I197d6486a3b126f616ffe2f7e3e669d922a1e337\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/09eadf955e50eb6a3797a0a8c32c56d5622b3835"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/09eadf955e50eb6a3797a0a8c32c56d5622b3835"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
