)]}'
{"id":"openstack%2Foslo.limit~715507","triplet_id":"openstack%2Foslo.limit~master~I7ca780a2cff32031c562482b804888e5d49712c1","project":"openstack/oslo.limit","branch":"master","topic":"update-hacking","hashtags":[],"change_id":"I7ca780a2cff32031c562482b804888e5d49712c1","subject":"Update hacking for Python3","status":"MERGED","created":"2020-03-27 18:15:10.000000000","updated":"2020-03-29 00:30:45.000000000","submitted":"2020-03-29 00:29:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715507-1585441773390-1fb5c97b","meta_rev_id":"1682b8281ae25f426da7daf29f4f70a4aa8a1578","_number":715507,"virtual_id_number":715507,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-29 00:29:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-27 20:26:20.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-28 23:06:27.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-28 23:06:27.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-27 20:26:20.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-28 23:06:27.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2020-03-29 00:29:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9562dda8e12bf8e11b4013e7b3df55c6d9671ab0","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-27 18:15:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ed8777757cfc5cdefd0fae71d195cddcd816e94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 19:42:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/91435e40b73343dcb3a3f726c9102bf1 : SUCCESS in 3m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e0b9711026145cf9f2444ec0f7183bd : SUCCESS in 1h 20m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/914a869da36348e9976025edc9e27f3f : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3042e630e5c4488790e6a1a6e6e3c5c8 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0985d9610cd4a54b6a143a351465619 : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/79fb52b67edd446da73e1351b3eca97e : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/633ba9ce0d284225b10b45f7d80c482e : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e36cc6112fbb4d32b7c7f6b411dda542 : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b24c4a2d2ac4db59fbda7106b13d22f : SUCCESS in 5m 06s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/f8428efbee4f47ffb5f0ef8b77c80b84 : SUCCESS in 33m 11s","accounts_in_message":[],"_revision_number":1},{"id":"62ce84290a44af069e53e3c9f2b0f49df7eddf41","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-27 19:54:52.000000000","message":"Topic set to update-hacking","accounts_in_message":[],"_revision_number":1},{"id":"db856ace298ab4f2a859b68a3f03b1f79816eb55","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-27 20:26:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e8b4eb9d5bbea8637e9077d69fbf32c0082d63a7","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-03-28 23:06:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9bf997667636ebdbb3def8cf4ebc676188def47c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 23:06:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2de8a1de38e1536ca62f97d5be6c67a6d7e7c61d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-29 00:29:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b6859c7e5a4c4a2289eeabb9bcaa2220 : SUCCESS in 2m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f2a2f5fbce442768cce483a1d090e3a : SUCCESS in 1h 21m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d229ed42ebe74ca3a2aeb1b31d28c47c : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9a6c2b027584bedad4694485873c7f5 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fec6c1af8db547599033b8024c0c5ca0 : SUCCESS in 3m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/064251ef0b494e15bce94823ac4e98d9 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93bafe33fe744deabbf398930487ffed : SUCCESS in 3m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d12679761f4343f5b7adfdc5fbc0455d : SUCCESS in 2m 44s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/cf9f38e0a63447908e8335be64409f68 : SUCCESS in 25m 49s","accounts_in_message":[],"_revision_number":1},{"id":"91059a961db562ceeaeee3d816104a6288911405","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-29 00:29:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1682b8281ae25f426da7daf29f4f70a4aa8a1578","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-29 00:30:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49ddc8ff7ca74dea850dbb5137b145eb : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4e0ff3b319d4adabbbe2a3cbce33d84 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"009fd6c7035ad96b125b041c147030e8596e22a2","revisions":{"009fd6c7035ad96b125b041c147030e8596e22a2":{"kind":"REWORK","_number":1,"created":"2020-03-27 18:15:10.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/07/715507/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/07/715507/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/07/715507/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/07/715507/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/07/715507/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/07/715507/1"}}},"commit":{"parents":[{"commit":"84ed5bc4d957623d644ae05d2419c869d1cbe129","subject":"Merge \"drop use of six\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/84ed5bc4d957623d644ae05d2419c869d1cbe129"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-27 18:13:57.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-27 18:14:46.000000000","tz":60},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 2.0 which\nsupports Python 3.\n\nBlacklist: W504 line break after binary operator\n\nChange-Id: I7ca780a2cff32031c562482b804888e5d49712c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/009fd6c7035ad96b125b041c147030e8596e22a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/009fd6c7035ad96b125b041c147030e8596e22a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
