)]}'
{"id":"openstack%2Foslo.limit~726930","triplet_id":"openstack%2Foslo.limit~master~I415fdef757e5f78f1c201c32489bc00a1f3cebd0","project":"openstack/oslo.limit","branch":"master","topic":"add_user_guide","hashtags":[],"change_id":"I415fdef757e5f78f1c201c32489bc00a1f3cebd0","subject":"Add user guide about how to add a new service","status":"MERGED","created":"2020-05-11 19:12:55.000000000","updated":"2020-06-22 16:06:55.000000000","submitted":"2020-06-22 16:04:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726930-1592841859905-5e2ae216","meta_rev_id":"85f0df546a5c7205ed301f6a16fcf6a3d703ed86","_number":726930,"virtual_id_number":726930,"owner":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-22 16:04:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-22 15:53:20.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-06-01 16:47:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-06-22 15:53:20.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"date":"2020-06-01 21:57:20.000000000","_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-19 20:50:57.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2020-05-19 20:51:05.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2020-05-19 20:51:09.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-05-25 13:55:04.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-06-01 16:47:08.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-01 21:57:20.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-06-22 15:53:20.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-22 16:04:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ffa700fc6da29154514564dbbe4c83421fe1420d","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2020-05-11 19:12:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5fad3d71339d313916ef0058197eeef5cbe013c2","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2020-05-11 19:14:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ba07c87cca91793533b57014815b0b875a9983d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 19:23:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/024ec887f33e4277926195882e493510 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/279e05f150654846a9140c642c589e25 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":2},{"id":"2a63025f80bf784539dacd329fe4c42d5da7852f","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-15 21:56:25.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3b9761d7246b8803012f863cc5206f4698fffcb2","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2020-05-16 11:23:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7841f887665c7d40048e1a71727356a413a9fda9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 11:29:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3e32c0bc027429a8f220ade617778a3 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1c8945cdf6c4ae7a8e910a22385092e : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":3},{"id":"f4fab0d7e29f7414582aa21408c88e89dccfbad0","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2020-05-16 11:29:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2362414b25e145b6495c9425c9935c4e2b6451ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 11:33:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9700e65f8af412b84d8295b4dccf213 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fb86047e8d541098cbf10a6790f593e : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":4},{"id":"5785ecc363c07ac059234809125b0121af802b64","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-19 20:50:57.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"4e2dfae0fd7ae65562943a0069897cd6a98fab3b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-25 11:56:31.000000000","message":"Patch Set 4: Code-Review+1\n\nLGTM I only +1\u0027ed to lets people with more experience than me on oslo.limit to take a look. \n\nIf these changes will be not commented/reviewed by other people in the following days then I\u0027ll +2.","accounts_in_message":[],"_revision_number":4},{"id":"1e2af84bd2d4ba8ac18d6436ba5454650211ec3c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-25 13:55:04.000000000","message":"Patch Set 4: Code-Review-1\n\n(19 comments)\n\nLots of grammar/typo nits to address but otherwise LGTM","accounts_in_message":[],"_revision_number":4},{"id":"f570762b46db2b50ce2af5cb3c986b7b12f141f6","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2020-05-30 12:34:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"93b34b0bf15a9a3c11e5a62c6a9e4a2de9fac55e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 12:38:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d988ae979824bb5b2bb84270a307406 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28a75d588e0443bdbf376fbecd336393 : FAILURE in 3m 56s","accounts_in_message":[],"_revision_number":5},{"id":"e7dba70a19fb524896304e8342abbba7f2c4097b","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2020-05-30 15:05:54.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"883e6e459802bb674057818f3ca949db471b8b65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 15:13:58.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66eddfdf06b24a578c70faf0f515790b : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcc472e440124f15be6b0f4145c4f8dc : FAILURE in 3m 16s","accounts_in_message":[],"_revision_number":6},{"id":"78f61c3d13cf328577093f387e8d96f5df4ce7ff","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2020-05-30 17:00:14.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a3e894650ea56eb6cffdaefda85b300a1c3121e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 17:04:00.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d981d74195e4ae2a2a277812754dbe2 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cd71c14a86146d79838321d33e64dc3 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":7},{"id":"c061bd4cf0aaf888d1b0cbe40c36f904a33507b0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-01 16:38:44.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"831c4b6feae94c70d53c7976ee1f5ec7ad0590ab","author":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"date":"2020-06-01 16:40:06.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e8d789ee9cfb61a598db9fa53bb14011db534fdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 16:44:36.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cd9ff5cf9cc4c32bb50ed6fb1e4f903 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8fc542624354ba8bf0b3125f81df801 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":8},{"id":"b23222acbdfd9f0b8807fd9a1262065a38052aca","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-01 16:47:08.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"386c79d2f709e18cd69f093523ba341e8f53c3d7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-22 15:53:20.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\nVery useful - thanks!","accounts_in_message":[],"_revision_number":8},{"id":"3c1327d825c8aa78eb4d8479e9732d8cab264d0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 15:54:44.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"6cd867059cdb456c7e532247a291a1bbb25b732a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 16:04:19.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8633198f0944540ab4ecd8b94f92ae8 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ba21a8d70b741879360deb1792d44e9 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":8},{"id":"f0ee9ebf297a4e319565758aa06cc40ef743c494","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 16:04:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"b0d90ce6b0e4475339ccc6fa9fc07d3f0b58d2f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 16:06:55.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/946c45bd7e8c47248bc4d4b450d79cbd : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"7f93f4607036cde28cca8a2200d91ea102007ae9","revisions":{"a76b7178e9bcd29f9ab5627c6e46d7c3ce1cb316":{"kind":"REWORK","_number":1,"created":"2020-05-11 19:12:55.000000000","uploader":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"ref":"refs/changes/30/726930/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/30/726930/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/1"}}},"commit":{"parents":[{"commit":"3b5a820db3cb51224c482123f2e40b74e460b5ab","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3b5a820db3cb51224c482123f2e40b74e460b5ab"}]}],"author":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-05-11 19:10:39.000000000","tz":0},"committer":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-05-11 19:10:39.000000000","tz":0},"subject":"Add user guide about how to add a new service","message":"Add user guide about how to add a new service\n\nThis commit add a new user guide about how a service\ndeveloper can easily integrate oslo_limit and use\nKeystone unified limit system.\n\nChange-Id: I415fdef757e5f78f1c201c32489bc00a1f3cebd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/a76b7178e9bcd29f9ab5627c6e46d7c3ce1cb316"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/a76b7178e9bcd29f9ab5627c6e46d7c3ce1cb316"}]},"branch":"refs/heads/master"},"048382da75cedd40975576a00302b5b4ca22e337":{"kind":"REWORK","_number":2,"created":"2020-05-11 19:14:01.000000000","uploader":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"ref":"refs/changes/30/726930/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/30/726930/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/2"}}},"commit":{"parents":[{"commit":"3b5a820db3cb51224c482123f2e40b74e460b5ab","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3b5a820db3cb51224c482123f2e40b74e460b5ab"}]}],"author":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-05-11 19:10:39.000000000","tz":0},"committer":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-05-11 19:13:48.000000000","tz":0},"subject":"Add user guide about how to add a new service","message":"Add user guide about how to add a new service\n\nThis commit add a new user guide about how a service\ndeveloper can easily integrate oslo_limit and use\nKeystone unified limit system.\n\nChange-Id: I415fdef757e5f78f1c201c32489bc00a1f3cebd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/048382da75cedd40975576a00302b5b4ca22e337"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/048382da75cedd40975576a00302b5b4ca22e337"}]},"branch":"refs/heads/master"},"c07861b0bf5963189be0e7724de5bd1447343e1b":{"kind":"REWORK","_number":3,"created":"2020-05-16 11:23:41.000000000","uploader":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"ref":"refs/changes/30/726930/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/30/726930/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/3"}}},"commit":{"parents":[{"commit":"3b5a820db3cb51224c482123f2e40b74e460b5ab","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3b5a820db3cb51224c482123f2e40b74e460b5ab"}]}],"author":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-05-11 19:10:39.000000000","tz":0},"committer":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-05-16 11:23:24.000000000","tz":0},"subject":"Add user guide about how to add a new service","message":"Add user guide about how to add a new service\n\nThis commit add a new user guide about how a service\ndeveloper can easily integrate oslo_limit and use\nKeystone unified limit system.\n\nChange-Id: I415fdef757e5f78f1c201c32489bc00a1f3cebd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/c07861b0bf5963189be0e7724de5bd1447343e1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/c07861b0bf5963189be0e7724de5bd1447343e1b"}]},"branch":"refs/heads/master"},"6e05904d82b06afb421d4f819e24c2891be2dd67":{"kind":"REWORK","_number":4,"created":"2020-05-16 11:29:40.000000000","uploader":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"ref":"refs/changes/30/726930/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/30/726930/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/4"}}},"commit":{"parents":[{"commit":"3b5a820db3cb51224c482123f2e40b74e460b5ab","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3b5a820db3cb51224c482123f2e40b74e460b5ab"}]}],"author":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-05-11 19:10:39.000000000","tz":0},"committer":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-05-16 11:28:51.000000000","tz":0},"subject":"Add user guide about how to add a new service","message":"Add user guide about how to add a new service\n\nThis commit add a new user guide about how a service\ndeveloper can easily integrate oslo_limit and use\nKeystone unified limit system.\n\nChange-Id: I415fdef757e5f78f1c201c32489bc00a1f3cebd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/6e05904d82b06afb421d4f819e24c2891be2dd67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/6e05904d82b06afb421d4f819e24c2891be2dd67"}]},"branch":"refs/heads/master"},"4777080e56fe7d6e48cfdb15ad5a2c284290e7bc":{"kind":"REWORK","_number":5,"created":"2020-05-30 12:34:10.000000000","uploader":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"ref":"refs/changes/30/726930/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/30/726930/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/5"}}},"commit":{"parents":[{"commit":"3b5a820db3cb51224c482123f2e40b74e460b5ab","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3b5a820db3cb51224c482123f2e40b74e460b5ab"}]}],"author":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-05-11 19:10:39.000000000","tz":0},"committer":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-05-30 12:34:00.000000000","tz":0},"subject":"Add user guide about how to add a new service","message":"Add user guide about how to add a new service\n\nThis commit add a new user guide about how a service\ndeveloper can easily integrate oslo_limit and use\nKeystone unified limit system.\n\nChange-Id: I415fdef757e5f78f1c201c32489bc00a1f3cebd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/4777080e56fe7d6e48cfdb15ad5a2c284290e7bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/4777080e56fe7d6e48cfdb15ad5a2c284290e7bc"}]},"branch":"refs/heads/master"},"30a567c52a34f16eb406026c2ff71debb3e5da14":{"kind":"REWORK","_number":6,"created":"2020-05-30 15:05:54.000000000","uploader":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"ref":"refs/changes/30/726930/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/30/726930/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/6"}}},"commit":{"parents":[{"commit":"3b5a820db3cb51224c482123f2e40b74e460b5ab","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3b5a820db3cb51224c482123f2e40b74e460b5ab"}]}],"author":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-05-11 19:10:39.000000000","tz":0},"committer":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-05-30 15:05:44.000000000","tz":0},"subject":"Add user guide about how to add a new service","message":"Add user guide about how to add a new service\n\nThis commit add a new user guide about how a service\ndeveloper can easily integrate oslo_limit and use\nKeystone unified limit system.\n\nChange-Id: I415fdef757e5f78f1c201c32489bc00a1f3cebd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/30a567c52a34f16eb406026c2ff71debb3e5da14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/30a567c52a34f16eb406026c2ff71debb3e5da14"}]},"branch":"refs/heads/master"},"59ea58ede7f191b9722bb680feb5c4d5ce3e0bfa":{"kind":"REWORK","_number":7,"created":"2020-05-30 17:00:14.000000000","uploader":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"ref":"refs/changes/30/726930/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/30/726930/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/7"}}},"commit":{"parents":[{"commit":"3b5a820db3cb51224c482123f2e40b74e460b5ab","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3b5a820db3cb51224c482123f2e40b74e460b5ab"}]}],"author":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-05-11 19:10:39.000000000","tz":0},"committer":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-05-30 17:00:06.000000000","tz":0},"subject":"Add user guide about how to add a new service","message":"Add user guide about how to add a new service\n\nThis commit add a new user guide about how a service\ndeveloper can easily integrate oslo_limit and use\nKeystone unified limit system.\n\nChange-Id: I415fdef757e5f78f1c201c32489bc00a1f3cebd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/59ea58ede7f191b9722bb680feb5c4d5ce3e0bfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/59ea58ede7f191b9722bb680feb5c4d5ce3e0bfa"}]},"branch":"refs/heads/master"},"7f93f4607036cde28cca8a2200d91ea102007ae9":{"kind":"REWORK","_number":8,"created":"2020-06-01 16:40:06.000000000","uploader":{"_account_id":28595,"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","username":"alistarle"},"ref":"refs/changes/30/726930/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/30/726930/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/30/726930/8"}}},"commit":{"parents":[{"commit":"3b5a820db3cb51224c482123f2e40b74e460b5ab","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3b5a820db3cb51224c482123f2e40b74e460b5ab"}]}],"author":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-05-11 19:10:39.000000000","tz":0},"committer":{"name":"Victor Coutellier","email":"victor.coutellier@gmail.com","date":"2020-06-01 16:39:57.000000000","tz":0},"subject":"Add user guide about how to add a new service","message":"Add user guide about how to add a new service\n\nThis commit add a new user guide about how a service\ndeveloper can easily integrate oslo_limit and use\nKeystone unified limit system.\n\nChange-Id: I415fdef757e5f78f1c201c32489bc00a1f3cebd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/7f93f4607036cde28cca8a2200d91ea102007ae9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/7f93f4607036cde28cca8a2200d91ea102007ae9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
