)]}'
{"id":"openstack%2Foslo.limit~728514","triplet_id":"openstack%2Foslo.limit~master~I9360028e96dd2b99fee1e31ebf9645e44e845d5f","project":"openstack/oslo.limit","branch":"master","hashtags":[],"change_id":"I9360028e96dd2b99fee1e31ebf9645e44e845d5f","subject":"Remove the unused coding style modules","status":"MERGED","created":"2020-05-15 17:10:37.000000000","updated":"2020-05-26 13:30:40.000000000","submitted":"2020-05-26 13:28:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728514-1590499715705-60d2867d","meta_rev_id":"1901736f7accd30cba15f65f45e8c6011ccbedc3","_number":728514,"virtual_id_number":728514,"owner":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-05-17 02:58:21.000000000","_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-26 13:28:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-26 12:14:29.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":2,"date":"2020-05-25 14:46:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-05-26 12:14:29.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-25 14:46:04.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-05-26 12:14:29.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-26 13:28:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3c7144f2e66666de2f97175e64c8ce771e035ae6","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-15 17:10:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23c06c054abad0678deb00bada0977cce79168a2","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-15 17:19:17.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d571dc7fd76302765bf146a2645f0a8c767112db","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-05-17 02:58:21.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"423ddeb1abc41e3e91de438846347d42ab17f6c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 04:26:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e3ea610792514adbb551ad77395fd4d6 : SUCCESS in 2m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b43c1cfa40004d648e300b6f32787554 : SUCCESS in 1h 23m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5fa31648c604fabaf89253ba8d56f6c : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/810db59825084d188efa0deb54843077 : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7e3fe969c32417da8a027d97dcc0e50 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d72d5cbed0a947ed8bffa7499421a15e : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29ec44e84fe340b88a9a0a2237f793ad : SUCCESS in 3m 27s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/ed8afe91814c47288aa2c37531d2cabe : SUCCESS in 28m 06s","accounts_in_message":[],"_revision_number":2},{"id":"3b9ce1266b69b6e1ac1afed8ef49a93258fa24e6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-25 14:46:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8c4aa7449dd8a43f6c28ae5ec705cd10256dad9d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-26 12:14:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"94cb060bcb9f6f5ad7f857d293774a5fe1cb0db6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 12:14:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"93f4537c7ee6a6c2732b192264f8b2a0a6e49ceb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 13:28:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f0ed793d4ae44c7b882d87c0aa9a8fb5 : SUCCESS in 5m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aed6a6a751c54da49c343dd2ada9c519 : SUCCESS in 1h 12m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86f3d28e86ac46de94ece4a48ba1359f : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a133c718882247d9898292318b04494a : SUCCESS in 9m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/276113be9a6844f1b76b60b04082f9d0 : SUCCESS in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3136c754e9fc4987a0a1d7e8b604fa9b : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75252557f1d3459c9d0180e5e8e0ac63 : SUCCESS in 7m 58s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/325ac42a505041ae80a9837ac8fdd656 : SUCCESS in 30m 52s","accounts_in_message":[],"_revision_number":2},{"id":"427d3f1557253dd5d01877ff423c1e3a6c384e8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 13:28:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1901736f7accd30cba15f65f45e8c6011ccbedc3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 13:30:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2ba946927d144cf90af0745b254d815 : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ee778fa233290f9ab0fa543ca82e8a88ee59fbfe","revisions":{"588c6d27f320d2aaf1cecb2bf880207286b0724d":{"kind":"REWORK","_number":1,"created":"2020-05-15 17:10:37.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/14/728514/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/14/728514/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/14/728514/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/14/728514/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/14/728514/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/14/728514/1"}}},"commit":{"parents":[{"commit":"3b5a820db3cb51224c482123f2e40b74e460b5ab","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3b5a820db3cb51224c482123f2e40b74e460b5ab"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-15 17:10:09.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-15 17:10:09.000000000","tz":480},"subject":"Remove the unused coding style modules","message":"Remove the unused coding style modules\n\nChange-Id: I9360028e96dd2b99fee1e31ebf9645e44e845d5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/588c6d27f320d2aaf1cecb2bf880207286b0724d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/588c6d27f320d2aaf1cecb2bf880207286b0724d"}]},"branch":"refs/heads/master"},"ee778fa233290f9ab0fa543ca82e8a88ee59fbfe":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-15 17:19:17.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/14/728514/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/14/728514/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/14/728514/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/14/728514/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/14/728514/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/14/728514/2"}}},"commit":{"parents":[{"commit":"3b5a820db3cb51224c482123f2e40b74e460b5ab","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3b5a820db3cb51224c482123f2e40b74e460b5ab"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-15 17:10:09.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-05-15 17:19:14.000000000","tz":0},"subject":"Remove the unused coding style modules","message":"Remove the unused coding style modules\n\nPython modules related to coding style checks (listed in blacklist.txt in\nopenstack/requirements repo) are dropped from lower-constraints.txt\nas they are not actually used in tests (other than pep8).\n\nmore info: https://github.com/openstack/requirements/blob/master/blacklist.txt\n\nChange-Id: I9360028e96dd2b99fee1e31ebf9645e44e845d5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/ee778fa233290f9ab0fa543ca82e8a88ee59fbfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/ee778fa233290f9ab0fa543ca82e8a88ee59fbfe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
