)]}'
{"id":"openstack%2Foslo.limit~728567","triplet_id":"openstack%2Foslo.limit~master~I5ec9dda8582f55fc0b287c422d5c43ad2e23b9b8","project":"openstack/oslo.limit","branch":"master","topic":"reno-openstackdocstheme","hashtags":[],"change_id":"I5ec9dda8582f55fc0b287c422d5c43ad2e23b9b8","subject":"Switch to newer openstackdocstheme and reno versions","status":"MERGED","created":"2020-05-15 19:54:57.000000000","updated":"2020-05-22 22:37:03.000000000","submitted":"2020-05-22 22:35:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728567-1590186953348-7d9b3224","meta_rev_id":"decd75bc3c3f63183e3f558cb56b78469342539b","_number":728567,"virtual_id_number":728567,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-22 22:35:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"all":[{"value":1,"date":"2020-05-22 06:19:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":2,"date":"2020-05-22 21:00:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-05-22 16:29:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":1,"date":"2020-05-22 21:00:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-22 06:19:20.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2020-05-22 16:29:58.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-05-22 21:00:48.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-05-22 22:35:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9c1584e425ef57a7c924da911988753d70f2c5e3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-15 19:54:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0cdd9798ce5de693a860ba68955c930911005cf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 21:34:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3633ac0d333149b5a67412bb781d5223 : SUCCESS in 3m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b665bfe411434fd180f5f06809a8d093 : SUCCESS in 1h 29m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eae6fa48de0b49beab8d380b451c87bf : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dc868fa533344668ea22707e7762ef3 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/449d1cf3f5a64fb890b8b3ae1e382132 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d2c23e0c29e433881034543fb512441 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bdc8c826df744b2bacede712bd4abb3 : SUCCESS in 4m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c17716319d8430da50ff3f27c32866e : SUCCESS in 3m 31s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/e0024144c32c459b9984d8e996c0e42d : SUCCESS in 30m 37s","accounts_in_message":[],"_revision_number":1},{"id":"f736a36b5267994edbf3c9138805826515736c22","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-18 16:34:38.000000000","message":"Topic changed from reno-openstacksdocstheme to reno-openstackdocstheme","accounts_in_message":[],"_revision_number":1},{"id":"e9fac93a116c1151f65a82109cafe9134b23f408","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-18 19:05:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1586649a0cf021ae6c02bcba38fef6b04a4e79d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 20:40:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/beb3ccb82d13430896ba7266d9b50d47 : SUCCESS in 2m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/69026b03a6134073aabf2f3620d438a2 : SUCCESS in 1h 30m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0dffb0045e844b3913551a9fd748f28 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66407b97e807411687cc68f5616095a7 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19c87bef7ff044749b3fbee155dabf81 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bdc97c340474d0baf727937c7fe8f05 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80b75cacfc2b40d59974deefb8d78dc4 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37933d6d780748559140b583b858e7a3 : SUCCESS in 4m 34s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/e8d667c672a74764a2d65346fe481318 : SUCCESS in 31m 34s","accounts_in_message":[],"_revision_number":2},{"id":"d2f5aee93b949226bd611c6c56b50bb9fec838f6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-19 12:49:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cf2787969997092b7b6eb3f09084453ce4a3a24d","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-20 06:25:47.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"dfaab6db801110858829902dc73820fbee0a176b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-21 07:24:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"286bc4b4747a1613fd504896f6e2b7f372ef723e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 08:47:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/904efea570744538a08f0c03e51f9ce8 : SUCCESS in 4m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89eb5aa8322f4e1c96002f7c1e4c5a62 : SUCCESS in 1h 22m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f039936331a4d9e9742766fdaff57d5 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/330d272f3e014670999240b8b27c6372 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cd02dee9fdc4bfe9570756e9f741c5c : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c292cf6138394227abd319c5ff14c615 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cac93c7ba9f474f9cd71b71b6e37ea8 : SUCCESS in 4m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/258549c6140c48b0952ffbaf981a62df : SUCCESS in 5m 06s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/89294df54e754f9fbcd58a6919edb08d : SUCCESS in 26m 07s","accounts_in_message":[],"_revision_number":3},{"id":"b8e08d0190d8f855cbf7c1bf153dcd8439b80942","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-21 13:30:30.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b22d39111627338e1fb93ebbc7a6d2c9606f3676","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-21 13:42:37.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"9b548d34b7b8679c28584f08af40ecd19e68529e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 15:05:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cf5e44b8a21a4d6dbe64f56b7e80981a : SUCCESS in 3m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/045093d1d8b34086aa7805051101d4dd : SUCCESS in 1h 20m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9e885971d3d48c3a160f00366ac5896 : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e80e1756f8dd4fae9e82a563bfed6628 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73e7139d284f4b6bb8fc47fd277baa76 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3457763721524523897d2c4234bd884e : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49b7c259e01b4acfb0c1db85e2404b91 : SUCCESS in 5m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a49099f436ff4fac844742612507cbd6 : SUCCESS in 3m 45s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/6d7208a679194b208ec1f93c30d94922 : SUCCESS in 30m 27s","accounts_in_message":[],"_revision_number":4},{"id":"5fc38e36d639d2904202da9318e21e35799b7ba9","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-22 06:19:20.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"019b9c8f44fb550e647813b57b04ee358a171d97","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-22 16:29:58.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"83373496ed750b9e41d1b479f6e7b32ca1e66bb1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-22 21:00:48.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"4acee6e06e4f4a62049bde12c322ddd1478aedc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 21:00:59.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"aa934214eed54666e9c196aac61a5b364b044cc1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 22:35:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"3648d5a38422be3407f220773897d85746d64514","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 22:35:53.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4f14a50870dd4bc898ae8e95687b4761 : SUCCESS in 2m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8bf23bc91c3b4a8f80c180b08ef45893 : SUCCESS in 1h 29m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee1eedd1b14544919c01b3a36add70cd : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa8b7ad9ed0a4f2282eb4096064f147e : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22ada1041ec1430b8aaed0efc19ec193 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a867275fcb3f499d9a0ffdc055c0f5d1 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/051872aa61bd4db3b8723bacef5bf4cc : SUCCESS in 3m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ac18b59f738449ab78a8f3ad8b91e23 : SUCCESS in 2m 57s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/048a9a87ff234d34a6971b03183396cc : SUCCESS in 28m 23s","accounts_in_message":[],"_revision_number":4},{"id":"decd75bc3c3f63183e3f558cb56b78469342539b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 22:37:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d8997cdb31744408a21b7b38c8ba636 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f5fe5f6021b4d7c856eda9e06933f54 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"16f909831205cce28af7a1e8a6eb65ce5d65d4f4","revisions":{"3237d288d892569ed019f5f5f9da573ad50377d7":{"kind":"REWORK","_number":1,"created":"2020-05-15 19:54:57.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/67/728567/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/67/728567/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/67/728567/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/67/728567/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/67/728567/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/67/728567/1"}}},"commit":{"parents":[{"commit":"3b5a820db3cb51224c482123f2e40b74e460b5ab","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3b5a820db3cb51224c482123f2e40b74e460b5ab"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-15 19:51:27.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-15 19:54:49.000000000","tz":120},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.1.2 and reno 3.1.0 versions. Using\nthese versions will allow parallelizing building of documents.\n\nUpdate Sphinx version as well.\n\nRemove doc requirments from lower-constraints, they are not used for\ninstall. Remove also hacking, the version is ancient and not needed\nthere either.\n\nopenstackdocstheme renames some variables, so follow the renames. A\ncouple of variables are also not needed anymore, remove them.\n\nSet openstackdocs_auto_name to use project as name.\n\nDepends-On: https://review.opendev.org/728432\nChange-Id: I5ec9dda8582f55fc0b287c422d5c43ad2e23b9b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3237d288d892569ed019f5f5f9da573ad50377d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3237d288d892569ed019f5f5f9da573ad50377d7"}]},"branch":"refs/heads/master"},"6797dd6bf5c3d072bb4e2334cd526871fa2e6de0":{"kind":"REWORK","_number":2,"created":"2020-05-18 19:05:11.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/67/728567/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/67/728567/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/67/728567/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/67/728567/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/67/728567/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/67/728567/2"}}},"commit":{"parents":[{"commit":"3b5a820db3cb51224c482123f2e40b74e460b5ab","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3b5a820db3cb51224c482123f2e40b74e460b5ab"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-15 19:51:27.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-18 19:04:41.000000000","tz":120},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.2.0 and reno 3.1.0 versions. Using\nthese versions will allow especially:\n* Linking from HTML to PDF document\n* Allow parallel building of documents\n* Fix some rendering problems\n\nUpdate Sphinx version as well.\n\nRemove doc requirments from lower-constraints, they are not used for\ninstall. Remove also hacking, the version is ancient and not needed\nthere either.\n\nopenstackdocstheme renames some variables, so follow the renames. A\ncouple of variables are also not needed anymore, remove them.\n\nSet openstackdocs_auto_name to use project as name.\n\nDepends-On: https://review.opendev.org/728938\nChange-Id: I5ec9dda8582f55fc0b287c422d5c43ad2e23b9b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/6797dd6bf5c3d072bb4e2334cd526871fa2e6de0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/6797dd6bf5c3d072bb4e2334cd526871fa2e6de0"}]},"branch":"refs/heads/master"},"eef4cbc62fa202190dff040c9847b352950de804":{"kind":"REWORK","_number":3,"created":"2020-05-21 07:24:23.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/67/728567/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/67/728567/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/67/728567/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/67/728567/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/67/728567/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/67/728567/3"}}},"commit":{"parents":[{"commit":"3b5a820db3cb51224c482123f2e40b74e460b5ab","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3b5a820db3cb51224c482123f2e40b74e460b5ab"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-15 19:51:27.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-21 07:24:18.000000000","tz":120},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.2.0 and reno 3.1.0 versions. Using\nthese versions will allow especially:\n* Linking from HTML to PDF document\n* Allow parallel building of documents\n* Fix some rendering problems\n\nUpdate Sphinx version as well.\n\nRemove doc requirments from lower-constraints, they are not used for\ninstall. Remove also hacking, the version is ancient and not needed\nthere either.\n\nopenstackdocstheme renames some variables, so follow the renames. A\ncouple of variables are also not needed anymore, remove them.\n\nSet openstackdocs_auto_name to use project as name.\n\nChange pygments_style to \u0027native\u0027 since old theme version always used\n\u0027native\u0027 and the theme now respects the setting and using \u0027sphinx\u0027 can\nlead to some strange rendering.\n\nSee also\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-May/014971.html\n\nChange-Id: I5ec9dda8582f55fc0b287c422d5c43ad2e23b9b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/eef4cbc62fa202190dff040c9847b352950de804"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/eef4cbc62fa202190dff040c9847b352950de804"}]},"branch":"refs/heads/master"},"16f909831205cce28af7a1e8a6eb65ce5d65d4f4":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-05-21 13:42:37.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/67/728567/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/67/728567/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/67/728567/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/67/728567/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/67/728567/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/67/728567/4"}}},"commit":{"parents":[{"commit":"3b5a820db3cb51224c482123f2e40b74e460b5ab","subject":"Add Python3 victoria unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3b5a820db3cb51224c482123f2e40b74e460b5ab"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-15 19:51:27.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-05-21 13:42:35.000000000","tz":0},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.2.1 and reno 3.1.0 versions. Using\nthese versions will allow especially:\n* Linking from HTML to PDF document\n* Allow parallel building of documents\n* Fix some rendering problems\n\nUpdate Sphinx version as well.\n\nRemove doc requirments from lower-constraints, they are not used for\ninstall. Remove also hacking, the version is ancient and not needed\nthere either.\n\nopenstackdocstheme renames some variables, so follow the renames. A\ncouple of variables are also not needed anymore, remove them.\n\nSet openstackdocs_auto_name to use project as name.\n\nChange pygments_style to \u0027native\u0027 since old theme version always used\n\u0027native\u0027 and the theme now respects the setting and using \u0027sphinx\u0027 can\nlead to some strange rendering.\n\nSee also\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-May/014971.html\n\nChange-Id: I5ec9dda8582f55fc0b287c422d5c43ad2e23b9b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/16f909831205cce28af7a1e8a6eb65ce5d65d4f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/16f909831205cce28af7a1e8a6eb65ce5d65d4f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
