)]}'
{"id":"openstack%2Foslo.limit~742915","triplet_id":"openstack%2Foslo.limit~master~I568123c93c32cb224628c4a8371a9a0e7c19cf67","project":"openstack/oslo.limit","branch":"master","topic":"bump-bandit","hashtags":[],"change_id":"I568123c93c32cb224628c4a8371a9a0e7c19cf67","subject":"Bump bandit version","status":"MERGED","created":"2020-07-24 15:33:47.000000000","updated":"2020-08-06 16:34:15.000000000","submitted":"2020-08-06 16:30:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742915-1596731435957-2e16cef6","meta_rev_id":"794c23005e0caa992b9ab0c58f4d8e5e0977b6a4","_number":742915,"virtual_id_number":742915,"owner":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-06 16:30:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-08-06 14:58:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-07-24 16:44:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-08-06 14:58:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-24 16:44:45.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-06 14:58:02.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-08-06 16:30:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a2efc48e94f13b4c9163815f3d0358d37f2523d8","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-07-24 15:33:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fdd305550a13e6c9255e6341867727fe1dbb5ace","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-24 16:44:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"85249c8e43016edd6e53953271977be09dd8066a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 17:14:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/97068f5db25b4cbe9ee17941e0f4e09a : SUCCESS in 4m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1856e571a8141d7bb1fc40c6c4b8ea0 : SUCCESS in 1h 30m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53a69532e53d40b9bbf92d0b641397ad : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1a6d96abcb845f3aadfa036e9617f8a : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/643252bd7dc74704bc84de9a2f1283b5 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc813014f58f4ed6895b4dd7af823586 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18dcb78220d54f0a8172dd7f802ee63b : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e357f1a1bfc4a72b718bd51893cab0d : SUCCESS in 4m 49s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/ad7565d2ea004aebb9c35e73ea48e8f6 : SUCCESS in 29m 39s","accounts_in_message":[],"_revision_number":1},{"id":"01d925918c47fa6fa7ad3bc3c8c9c3dbc1e66cc5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-06 14:58:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9266b6857fa6fd737b38153fba8516d414fb665f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 14:58:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"45d20954b3777d624364e351ad40d67c8d7de15f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 16:30:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/87e1df170e674e63a61dcb5d1e7d0535 : SUCCESS in 3m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/514d87ec12d14775b2bb96263ece5029 : SUCCESS in 1h 24m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84912e135c5845e7b498159a623cd839 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6ceab3a280840808e5253e275c2554e : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12f54c5d6e974903b14b9d51800abf7a : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3a20b230bdf4dcba8c1a4bd8f9f7206 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32733337b2f54deebd3c832b12ae4925 : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcba89ecc36f4f30859c8458c5ec07f6 : SUCCESS in 3m 46s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/c26f3a7cee2b4b1a965dde717df9d285 : SUCCESS in 32m 49s","accounts_in_message":[],"_revision_number":1},{"id":"245319c623e55cfe7d85d0230b41aa27bb18741e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 16:30:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e9f9b9f7e3a0eaf1eec6003006445c7f330467de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 16:34:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edfd3ca3409d42108e80c7e5c918df53 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cd60b86eef844dfa952c38ef828d174 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ddb62b9fd41380870ee5ab2d46ca11a33fbfd990","revisions":{"ddb62b9fd41380870ee5ab2d46ca11a33fbfd990":{"kind":"REWORK","_number":1,"created":"2020-07-24 15:33:47.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/15/742915/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/15/742915/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/15/742915/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/15/742915/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/15/742915/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/15/742915/1"}}},"commit":{"parents":[{"commit":"8fa46cdf509cec982cf95ec17729022d1a21fa3f","subject":"Merge \"Align contributing doc with oslo\u0027s policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/8fa46cdf509cec982cf95ec17729022d1a21fa3f"}]}],"author":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2020-07-24 15:33:41.000000000","tz":120},"committer":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2020-07-24 15:33:41.000000000","tz":120},"subject":"Bump bandit version","message":"Bump bandit version\n\nThis patch bumps bandit allowed version to \u003e\u003d1.6.0,\u003c1.7.0 in order to\navoid the errors detailed here https://github.com/PyCQA/bandit/pull/393\n\nChange-Id: I568123c93c32cb224628c4a8371a9a0e7c19cf67\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/ddb62b9fd41380870ee5ab2d46ca11a33fbfd990"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/ddb62b9fd41380870ee5ab2d46ca11a33fbfd990"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
