)]}'
{"id":"openstack%2Foslo.limit~753522","triplet_id":"openstack%2Foslo.limit~master~Ia65109bbb683489ba9a880d3a41baca5ef560d77","project":"openstack/oslo.limit","branch":"master","hashtags":[],"change_id":"Ia65109bbb683489ba9a880d3a41baca5ef560d77","subject":"Remove install unnecessary packages","status":"MERGED","created":"2020-09-23 06:51:59.000000000","updated":"2020-09-23 16:30:12.000000000","submitted":"2020-09-23 16:22:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753522-1600878172462-365cc730","meta_rev_id":"a9eaedd837317519193bf74ad3f0375e1bf9e62c","_number":753522,"virtual_id_number":753522,"owner":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-09-23 13:23:35.000000000","_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-23 16:22:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},{"value":2,"date":"2020-09-23 13:32:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-09-23 13:25:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},{"value":1,"date":"2020-09-23 13:32:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-23 13:25:15.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-23 13:32:04.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-09-23 16:22:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3d3d6cca413bc306c18ca857b720e70cd30ecb37","author":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"date":"2020-09-23 06:51:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"178a26a72c233a609f1afbce726766b9b8060f47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 08:20:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76def955ce744552a401e393a98b5ce8 : SUCCESS in 1h 18m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/740682a12ae74e2a981a31306506948f : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a239e5bace5447b9a16af7420704df1 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ed03cf9c5584dfca0a43fe1b2f3d343 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98489b8f75f5489a90648dd99bac7121 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/224ddd6176f1427094940c5f0ab84e75 : SUCCESS in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a75be518a4a4a5b8c26f918acfb298f : SUCCESS in 5m 02s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/85202679baf64bf78c7cd9dd3c2eae6c : SUCCESS in 28m 46s","accounts_in_message":[],"_revision_number":1},{"id":"9fbfe19dd8306430c32897d7047971760c1724a6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-09-23 09:21:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9972f252c0b0f01b0abb6fc93bdfdc375d74fd03","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-09-23 09:23:34.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f1fec5daf44b082e8510ae856387217af383d493","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-09-23 09:25:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8894da28cb2cecc777db1feae7ff5b47489b7e83","author":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"date":"2020-09-23 13:21:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"44c40e2a7f2f49e77c4b99bc8d31373cd87d7cb4","author":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"date":"2020-09-23 13:23:35.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nThanks for your review, Stephen Finucane!","accounts_in_message":[],"_revision_number":2},{"id":"a1cdf67333f8e2ab2b6f6371c17456c68f3371a2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-23 13:25:15.000000000","message":"Patch Set 2: Code-Review+2\n\nStephen\u0027s comments seems addressed, LGTM","accounts_in_message":[],"_revision_number":2},{"id":"9a88ffaf424597ba9393e26366fb6537315913a4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-09-23 13:32:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nAssuming CI is happy","accounts_in_message":[],"_revision_number":2},{"id":"68ee2b385fd637778416f98f9f38f9d0e8a2a100","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 15:05:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/641c2949320b491a93235fb82f052a50 : SUCCESS in 1h 21m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a1197fb077e4064a0ad9cb05ef19fc2 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7ce87a959c84f6ab685e1fd485e6820 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa626c755ee3406983d20086ab11b68e : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/102cd5938db94a9bb786500590b1a5aa : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04fe06c68bc345ed8be2e822ec2f6cbc : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b80f66a7392d4ac9b1baf979fe90d914 : SUCCESS in 3m 06s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/d93949da29ad44b78e8f5713ba10ba13 : SUCCESS in 28m 54s","accounts_in_message":[],"_revision_number":2},{"id":"7ac8425d1fa7b385738055f54308603be17dedff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 15:05:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"77863dba478d30de8dfd55727fe694743a21f12f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 16:22:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"deca32f4523115e957dc57971027f5ba619b3ed0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 16:22:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/145ebb53efc74436afc26276fb4f1c9b : SUCCESS in 1h 14m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e53ebf8f5e540968b80b86687a0c84e : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d8974828e5146e48863e475e47c216c : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c75b35570914fea8c20ff5df0b45878 : SUCCESS in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a899c2ad8e5e4193ac5d8b32eab183d4 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca5c2773e259441cbd69bb04e050f70d : SUCCESS in 3m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51924fb9b5334d77a2160d2b426a0208 : SUCCESS in 3m 37s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/1fa301de320140e797daa5750d243d95 : SUCCESS in 30m 04s","accounts_in_message":[],"_revision_number":2},{"id":"a9eaedd837317519193bf74ad3f0375e1bf9e62c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 16:30:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/448410a614544d57a6f990f45df3792c : SUCCESS in 2m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6d0e12176034a2389a96c92b07a0473 : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9558a46c9494293e3faf12cdb70d41193f6af781","revisions":{"7100981a844a2f43e7b4ef441f3c4c4fe1414120":{"kind":"REWORK","_number":1,"created":"2020-09-23 06:51:59.000000000","uploader":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"ref":"refs/changes/22/753522/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/22/753522/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/22/753522/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/22/753522/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/22/753522/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/22/753522/1"}}},"commit":{"parents":[{"commit":"55675ba4473907ebd6134b55c0cefd178f7bbf48","subject":"Merge \"Adding pre-commit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/55675ba4473907ebd6134b55c0cefd178f7bbf48"}]}],"author":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2020-09-23 06:50:49.000000000","tz":480},"committer":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2020-09-23 06:51:53.000000000","tz":480},"subject":"Remove install unnecessary packages","message":"Remove install unnecessary packages\n\nThe docs requirements migrated to doc/requirements.txt\nwe need not install things from requirements.txt.\n\nChange-Id: Ia65109bbb683489ba9a880d3a41baca5ef560d77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/7100981a844a2f43e7b4ef441f3c4c4fe1414120"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/7100981a844a2f43e7b4ef441f3c4c4fe1414120"}]},"branch":"refs/heads/master"},"9558a46c9494293e3faf12cdb70d41193f6af781":{"kind":"REWORK","_number":2,"created":"2020-09-23 13:21:19.000000000","uploader":{"_account_id":31027,"name":"MaAoyu","email":"523354092@qq.com","username":"maaoyu"},"ref":"refs/changes/22/753522/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/22/753522/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/22/753522/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/22/753522/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/22/753522/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/22/753522/2"}}},"commit":{"parents":[{"commit":"55675ba4473907ebd6134b55c0cefd178f7bbf48","subject":"Merge \"Adding pre-commit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/55675ba4473907ebd6134b55c0cefd178f7bbf48"}]}],"author":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2020-09-23 06:50:49.000000000","tz":480},"committer":{"name":"maaoyu","email":"maaoyu@inspur.com","date":"2020-09-23 13:20:57.000000000","tz":480},"subject":"Remove install unnecessary packages","message":"Remove install unnecessary packages\n\nThe docs requirements migrated to doc/requirements.txt\nwe need not install things from requirements.txt.\n\nChange-Id: Ia65109bbb683489ba9a880d3a41baca5ef560d77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/9558a46c9494293e3faf12cdb70d41193f6af781"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/9558a46c9494293e3faf12cdb70d41193f6af781"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
