)]}'
{"id":"openstack%2Foslo.limit~760345","triplet_id":"openstack%2Foslo.limit~master~I64843401f516ba362943ec83117be70ab48803df","project":"openstack/oslo.limit","branch":"master","topic":"tox","hashtags":[],"change_id":"I64843401f516ba362943ec83117be70ab48803df","subject":"Update URLs in tox docs","status":"ABANDONED","created":"2020-10-29 11:56:39.000000000","updated":"2020-10-29 14:13:06.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9581a5c1d09b1d92d4308f356062ecec8e2435ea","_number":760345,"virtual_id_number":760345,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-29 13:40:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":-1,"date":"2020-10-29 13:50:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":-1,"date":"2020-10-29 12:19:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-29 12:19:08.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-29 13:40:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-29 13:50:17.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"735872c3540401c97ee78289f3c28ee2e1269987","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-10-29 11:56:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a05f9c49f09ace7e24bac4e158b17bb054d12bc7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-29 11:59:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9281038234da02f703051a66f3a10e08f0ae37a4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-29 12:00:15.000000000","message":"Patch Set 1: Code-Review-1\n\nThese changes aren\u0027t needed","accounts_in_message":[],"_revision_number":1},{"id":"cb45f093752aa52f938a4d321af28072ae611c13","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-10-29 12:04:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2bdde4ce19c45da954097a7804dae009cef99407","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-29 12:19:08.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"270cb419a6111e602aba78dfa584f6af593a0758","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 13:40:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9339362ee29443d189d297dde65878d9 : SUCCESS in 1h 26m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abfc6e9be3344b06bfd8607c896430a6 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc8d6aece8fb42afbb13b268d2bfa607 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ec162b1c28848a9a74b4ab7e2bb4eb5 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ce47cc5f64a46faa5c313353c2f39e6 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2570ff7b8962446c88761ba148d389f2 : SUCCESS in 3m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6820f3ad70ea43338f924153b91cade8 : SUCCESS in 3m 02s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/6141fe1537cf4ea3ae469191da883791 : SUCCESS in 52m 23s","accounts_in_message":[],"_revision_number":2},{"id":"9c2e2d8c3cadbe0ffdd996a901d9dc0d578740dc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-10-29 13:50:17.000000000","message":"Patch Set 2: Code-Review-1\n\nyeah, this is correct atm","accounts_in_message":[],"_revision_number":2},{"id":"0879ef6a17cea86a514c8816bac880a320c960e3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-10-29 14:13:06.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1fdedea023fce98196d89e757ff99d88f9c15d63","revisions":{"2fa3c97f1bdebef9fa54e440404f92ac418d0fd0":{"kind":"REWORK","_number":1,"created":"2020-10-29 11:56:39.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/45/760345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/45/760345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/45/760345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/45/760345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/45/760345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/45/760345/1"}}},"commit":{"parents":[{"commit":"592a0d09dd136fc0b00c18ed32d095e4d8203613","subject":"Fix hacking min version to 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/592a0d09dd136fc0b00c18ed32d095e4d8203613"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-10-29 11:55:40.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-10-29 11:55:40.000000000","tz":480},"subject":"Update requirements URLs in tox config","message":"Update requirements URLs in tox config\n\nUpdate the URL to the upper-constraints file\n\nChange-Id: I64843401f516ba362943ec83117be70ab48803df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/2fa3c97f1bdebef9fa54e440404f92ac418d0fd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/2fa3c97f1bdebef9fa54e440404f92ac418d0fd0"}]},"branch":"refs/heads/master"},"1fdedea023fce98196d89e757ff99d88f9c15d63":{"kind":"REWORK","_number":2,"created":"2020-10-29 12:04:29.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/45/760345/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/45/760345/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/45/760345/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/45/760345/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/45/760345/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/45/760345/2"}}},"commit":{"parents":[{"commit":"592a0d09dd136fc0b00c18ed32d095e4d8203613","subject":"Fix hacking min version to 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/592a0d09dd136fc0b00c18ed32d095e4d8203613"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-10-29 11:55:40.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-10-29 12:03:38.000000000","tz":480},"subject":"Update URLs in tox docs","message":"Update URLs in tox docs\n\nUpdate the URL to the upper-constraints file\n\nChange-Id: I64843401f516ba362943ec83117be70ab48803df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/1fdedea023fce98196d89e757ff99d88f9c15d63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/1fdedea023fce98196d89e757ff99d88f9c15d63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
