)]}'
{"id":"openstack%2Foslo.limit~761141","triplet_id":"openstack%2Foslo.limit~master~Idb0f7ff31a8fa788c940d70eddd1e7218691a31d","project":"openstack/oslo.limit","branch":"master","topic":"tox-python38","hashtags":[],"change_id":"Idb0f7ff31a8fa788c940d70eddd1e7218691a31d","subject":"Use py3 as the default runtime for tox","status":"MERGED","created":"2020-11-03 13:31:19.000000000","updated":"2020-11-04 11:49:47.000000000","submitted":"2020-11-04 11:48:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761141-1604490510241-8dccae41","meta_rev_id":"fe574d47c0b00c39de3a907ce5b371aaf84c1d10","_number":761141,"virtual_id_number":761141,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2020-11-04 08:53:35.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-04 11:48:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-11-03 15:26:49.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-11-03 16:00:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-11-03 16:00:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-03 15:26:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-11-03 16:00:58.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-11-04 11:48:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ebaea730fd8e3f9f80fc0da86127c78bcddf577e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-03 13:31:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63ab01227469502179b80a4cc9e85d6b2c7a2967","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-11-03 13:58:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d33185677e381be2fce442eb3bff16e4f4aa572a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-11-03 14:30:07.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nActually, maybe we want to use py3 (suggested by smcginnis) to avoid this tedious work every cycle?","accounts_in_message":[],"_revision_number":1},{"id":"c3cd55d690bf27d2b2e6ddcdf98766e1f7a8a5d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 15:04:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb959e3656e54315b0969525090d00cc : SUCCESS in 1h 19m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14e4bae148df49e18de23814660619f4 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cac231f4193a40aba6a19194df9018c7 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60aad42ca9b74d9eb5031771048bec49 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fbcef5408d44680aa3c0758eecc6001 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29d8a181aba24f9fbdc358b7fb8b6dca : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ecebd04e4de4c02b1e8d8a850a95f5e : SUCCESS in 4m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24f05549841940fe8dbff481e84e6b60 : SUCCESS in 3m 18s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/646421a002d340a180ff6cc3f67dbb5d : SUCCESS in 25m 28s","accounts_in_message":[],"_revision_number":1},{"id":"2a83fb09abf34a467eedbac2d11d44a887c8be21","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-03 15:13:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"671669e25872a69bd5d26beee40fc0e4e915ece1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-11-03 15:26:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"425830020b2a566caada717b04106c8c9e347629","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-11-03 16:00:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b3c7ae0d7fff290c4feded0d3e5f13de30cbc391","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 16:39:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9fe33e8316fc487ca3a9610fa21b37eb : SUCCESS in 1h 16m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea0a309c102c4295824229ef2a465a22 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86e9b6eb012a4aa6a0d841b36b417cbd : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b97db9cf65c2469dbb7ae78cbe9c7645 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54dcd6adae0e4272949cea0796253734 : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4487bc7cae51480f87b5f102c90d2386 : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c76b7ba47dba42628d676bc4de550a13 : SUCCESS in 3m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/294abb6cafb843f694fd584c5c53732b : SUCCESS in 3m 14s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/812e9e8207aa45e6a9e92c906d5609d4 : SUCCESS in 29m 56s","accounts_in_message":[],"_revision_number":2},{"id":"65685d8d58d1a4da6bca5cc97d5c27d9921def46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 16:39:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"52a3b2963d0e0a8fe5359db340325a95dc764cc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 18:03:12.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/732ea610b136459297e42beaf6f74e1b : SUCCESS in 1h 17m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54ff274d3702464fb8d8c3e2b8a4a0a0 : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a22875d8cd7d47bebdb8205140d660fc : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d159296060e46d491c01c1592722c0b : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a320b6bc56c04c7a9850438e0389e390 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/427f3fe5febe46a0842a859463e1e799 : SUCCESS in 4m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f08e1676929f4090a001a06587134a6a : SUCCESS in 2m 50s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/9fb3a8c14ec349f6a6c10c8432469aca : FAILURE in 3m 56s","accounts_in_message":[],"_revision_number":2},{"id":"08df167c9ea5fd1aa1b5e378e66928c12236d952","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-04 08:53:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"31229643f6e389d93643f171661a1f332f9d09e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 10:17:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bfe9fbc762bb4fcd809f637fa85a08cd : SUCCESS in 1h 15m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30a3bb2765ce460eb7c2a74e389fd74e : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a18adbc071394cc3b47b5ad290b6d07c : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85681b4be45a42eaa7a8bb195272698a : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5550fa629f4247a9ba7f508e109434b6 : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f8c374a448143eca73c0081687e7999 : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cc228cace0b4a39850f075310ac6020 : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48660e57e0b64045aef379fc08545faf : SUCCESS in 3m 02s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/5ed059f3ee5540abaf46e9210f0a304d : SUCCESS in 31m 17s","accounts_in_message":[],"_revision_number":2},{"id":"2108e5db9eb09e5205efdfbdd92d0d1221ea894b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 10:17:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d2a7c8d24fe8242ae94dfb092155894ff42db006","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 11:48:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9eef2b7bf58e47bc998655c45e093aaa : SUCCESS in 1h 24m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db17ce616db34c66a087bea3e491e3ed : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62e813c2c3be4fa888e8fcacb064dd22 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25fd9904f4d24450ac839bb205fbd266 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/399915e83c4f4ff9b69378339d52635e : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ed66f3fafa14da0800e627129474be8 : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00b5e7ff8ae142edb349579e5a8e0087 : SUCCESS in 2m 58s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/b1cb46bf992d4bfc85e8ab4404f5b0e8 : SUCCESS in 31m 44s","accounts_in_message":[],"_revision_number":2},{"id":"6ca3e197b58712a770249ae298d7df7763562db2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 11:48:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"09e7f23212979dc8d42574172344561db7df967d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 11:49:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/138e507b3b464f3c835b773484aed665 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42af902ba4124730b3428e9c0bc20632 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a5c678ce8b51ec76eefaa66f21752504ff42d2a4","revisions":{"5785e39244e596e64d756ca7afb5054caf73a200":{"kind":"REWORK","_number":1,"created":"2020-11-03 13:31:19.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/41/761141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/41/761141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/41/761141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/41/761141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/41/761141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/41/761141/1"}}},"commit":{"parents":[{"commit":"592a0d09dd136fc0b00c18ed32d095e4d8203613","subject":"Fix hacking min version to 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/592a0d09dd136fc0b00c18ed32d095e4d8203613"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-11-03 13:31:13.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-11-03 13:31:13.000000000","tz":60},"subject":"Use Python 3.8 as the default runtime for tox","message":"Use Python 3.8 as the default runtime for tox\n\nWallaby support officially the following runtimes [1]:\n- Python 3.6\n- Python 3.8\n\nDuring Victoria Python 3.7 was used as the default runtime [2] however this\nversion isn\u0027t longer officially supported.\n\nMoving on the most recent officially supported runtime as the default runtime\nfor tox.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/wallaby.html#python-runtimes-for-wallaby\n[2] https://governance.openstack.org/tc/reference/runtimes/victoria.html#python-runtimes-for-victoria\n\nChange-Id: Idb0f7ff31a8fa788c940d70eddd1e7218691a31d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/5785e39244e596e64d756ca7afb5054caf73a200"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/5785e39244e596e64d756ca7afb5054caf73a200"}]},"branch":"refs/heads/master"},"a5c678ce8b51ec76eefaa66f21752504ff42d2a4":{"kind":"REWORK","_number":2,"created":"2020-11-03 15:13:09.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/41/761141/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/41/761141/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/41/761141/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/41/761141/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/41/761141/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/41/761141/2"}}},"commit":{"parents":[{"commit":"592a0d09dd136fc0b00c18ed32d095e4d8203613","subject":"Fix hacking min version to 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/592a0d09dd136fc0b00c18ed32d095e4d8203613"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-11-03 13:31:13.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-11-03 15:12:44.000000000","tz":60},"subject":"Use py3 as the default runtime for tox","message":"Use py3 as the default runtime for tox\n\nMoving on py3 as the default runtime for tox to avoid to update this at\neach new cycle.\n\nWallaby support officially the following runtimes [1]:\n- Python 3.6\n- Python 3.8\n\nDuring Victoria Python 3.7 was used as the default runtime [2] however this\nversion isn\u0027t longer officially supported.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/wallaby.html#python-runtimes-for-wallaby\n[2] https://governance.openstack.org/tc/reference/runtimes/victoria.html#python-runtimes-for-victoria\n\nChange-Id: Idb0f7ff31a8fa788c940d70eddd1e7218691a31d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/a5c678ce8b51ec76eefaa66f21752504ff42d2a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/a5c678ce8b51ec76eefaa66f21752504ff42d2a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
