)]}'
{"id":"openstack%2Foslo.limit~761344","triplet_id":"openstack%2Foslo.limit~master~I1a006cf4368cc18c34dd6f236ecfe966e6b09388","project":"openstack/oslo.limit","branch":"master","topic":"tox_constraints_file","hashtags":[],"change_id":"I1a006cf4368cc18c34dd6f236ecfe966e6b09388","subject":"Use TOX_CONSTRAINTS_FILE","status":"MERGED","created":"2020-11-04 09:48:27.000000000","updated":"2021-04-16 13:12:31.000000000","submitted":"2021-04-16 13:10:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"761344-tox_constraints_file","meta_rev_id":"182520b3b40e4f85340a95f89c29fd2a3ff61821","_number":761344,"virtual_id_number":761344,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-16 13:10:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-04-16 10:24:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-04-16 10:24:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-04 15:48:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-11 15:36:32.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"03dea5cc78eaa4fe18750049726b64bf31f73c2a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-04 09:48:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ad36e877af5e3112bc2ece79a567c3d19950470","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 11:50:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f7b10b611e649d5908e8f4edcdc3853 : FAILURE in 1h 46m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cad3cffda4d144138c4b9b95b1cc0b1f : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99da869570294084a5f9143ae88ad50e : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/229ac19709024442a9415db82fc573e2 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31011ba47b424e3db4d03bf3888662e0 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f4ccd67f2c54c4fb1176bc1f1978b67 : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5bf882ec282491d8afd8d28c9a49024 : SUCCESS in 3m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05ce8da1c3854f12b73f77d8d36f6dcc : SUCCESS in 3m 12s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/d02e9a0bd17a4814ad1b8b7e886bf50c : SUCCESS in 27m 09s","accounts_in_message":[],"_revision_number":1},{"id":"27f1b93dcc5cd95c4fd342c28a9c4da8d299f809","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-04 13:59:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7c4214c0a33de891d64d366b6431ca5648b9f06e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 15:48:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1da7bad314304aea92b6f3930fe7901d : SUCCESS in 1h 29m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de10f4e4ab734bc3942172a17990932f : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/029be84f8d1149faa203e284fc88e89b : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/759b8516dc5945f387ad0cc0e44c0437 : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ac64c19af3a44f69c5c82fe4edcab8e : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c8e250dbfce4fcca610aa7c3f9e64e5 : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d46fa231a8b4d68b805f62eecc9ca58 : SUCCESS in 3m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b3354a49bad40e58cd3ed00dcc72666 : SUCCESS in 3m 28s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/6aa4547cbeab461f97ae7e1c0356c9cd : SUCCESS in 30m 29s","accounts_in_message":[],"_revision_number":1},{"id":"e75c41576c0cec245a044b85f33ad6ecec230451","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-12-11 15:36:32.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"734e25fac0d0f1aa47243289bc04a7fcb01cf28d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-14 09:30:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1cb9e978632c1d27848039b773fa95e00ff55f66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 11:59:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b03702cd91a4f1d86513c3939818fab : SUCCESS in 1h 50m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d48e135c2227445da02804d1cfc190e5 : FAILURE in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb82a72fc96649ff8461278684ecda77 : FAILURE in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/801bcaa8ec6c473586a49177c261ea6a : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19d4ed4ef3a043b292a263d3f8b64ee2 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6fa268bae34445ca2df8b6eb7623553 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a304d873a9a049cbbe1849d7f052c94d : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f3213ef7a014e8fb5bc3719e6882a94 : SUCCESS in 2m 57s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/046a1d2a2107473d8bc0b84217042b30 : SUCCESS in 34m 22s","accounts_in_message":[],"_revision_number":2},{"id":"df1f07e807ddaeb8ca691bb4c0cf30dabd526b39","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-22 08:11:49.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c823b36b7535499addd4215af201e60fa68b69b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 09:44:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4484a2d22bce4adf9b0ac49e2ad8bfe5 : SUCCESS in 1h 29m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0e5c4e76c4b4fcfba06ba474cedbc82 : FAILURE in 2m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8e19cfcb55e475b9b4934f8992ba357 : FAILURE in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3e2b394307041bba1459df65173f74f : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/592d223c05e8498c82efe52e2cc4a7a1 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04955c7306f84eadb4ceee8088ce59e2 : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/362d3f9c26514cb7bac5979c34e1b3c6 : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89eeff6662dd4389ac0efb6d691f8296 : SUCCESS in 3m 53s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/561e9974084643bf88b1549c9f45b8d6 : SUCCESS in 29m 11s","accounts_in_message":[],"_revision_number":2},{"id":"ab4658d883e1cbb82eceaaf8aeb327b600094895","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-16 10:24:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6c8bb31137927a5fabe7645f5f84d2a7998fca55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 11:35:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85f8695a93c147f586d04bd69c27c6ab : SUCCESS in 1h 09m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d46eb017e5374b188d555530c2dc34d6 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4879499db6224cfbaba8bb10d65be9b5 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b0ef440c9f544129e07e8dc54787f1c : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/342e310261d74332a622aaffeab0ad13 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3c863773e224c05a4cfa71377fbb45e : SUCCESS in 3m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b79051ba0ab84ddb843924be344ba822 : SUCCESS in 3m 52s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/5e29b17adf004a1abf0cfeffae870ce5 : SUCCESS in 19m 55s","accounts_in_message":[],"_revision_number":2},{"id":"e7d5b5f672dd355bf855d699ab7e26e3562b7f66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 11:36:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d52d4d03cb63e85716b6f54c4995fbb51412fa61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 13:10:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eef6343acd05407fabf034d239ab07f4 : SUCCESS in 1h 30m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f36cc3fbfe0455d92f626893837da59 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/443bf919e7d849c58ffc2bee5752d4af : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12e902858df64785a09382a41de753ed : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/188f2502bd0d489582fed6256b96ebbc : SUCCESS in 4m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9296a394420440c3993dc13b975ad212 : SUCCESS in 3m 51s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/4e8e46213778415eb63fce85f55277ab : SUCCESS in 30m 13s","accounts_in_message":[],"_revision_number":2},{"id":"5a6e2f235d140d66e0b2fc79330431980004842b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 13:10:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"182520b3b40e4f85340a95f89c29fd2a3ff61821","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 13:12:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23eac277ad104acd9734e90d3643d863 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7058488db81e42798b209b66cc20159e : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fc270afcc55df6471637fe4cd1f0295245bcf707","revisions":{"06425aef4e5aecff2134f15b36781510a78967aa":{"kind":"REWORK","_number":1,"created":"2020-11-04 09:48:27.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/44/761344/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/44/761344/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/44/761344/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/44/761344/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/44/761344/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/44/761344/1"}}},"commit":{"parents":[{"commit":"592a0d09dd136fc0b00c18ed32d095e4d8203613","subject":"Fix hacking min version to 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/592a0d09dd136fc0b00c18ed32d095e4d8203613"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-11-04 09:48:20.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-11-04 09:48:20.000000000","tz":60},"subject":"Use TOX_CONSTRAINTS_FILE","message":"Use TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\nThis allows to use upper-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\n[1] https://review.opendev.org/#/c/722814/\n[2] https://zuul-ci.org/docs/zuul-jobs/python-roles.html#rolevar-tox.tox_constraints_file\n\nChange-Id: I1a006cf4368cc18c34dd6f236ecfe966e6b09388\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/06425aef4e5aecff2134f15b36781510a78967aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/06425aef4e5aecff2134f15b36781510a78967aa"}]},"branch":"refs/heads/master"},"fc270afcc55df6471637fe4cd1f0295245bcf707":{"kind":"REWORK","_number":2,"created":"2021-01-14 09:30:52.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/44/761344/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/44/761344/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/44/761344/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/44/761344/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/44/761344/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/44/761344/2"}}},"commit":{"parents":[{"commit":"592a0d09dd136fc0b00c18ed32d095e4d8203613","subject":"Fix hacking min version to 3.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/592a0d09dd136fc0b00c18ed32d095e4d8203613"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-11-04 09:48:20.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-14 09:30:43.000000000","tz":60},"subject":"Use TOX_CONSTRAINTS_FILE","message":"Use TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\nThis allows to use upper-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\n[1] https://review.opendev.org/#/c/722814/\n[2] https://zuul-ci.org/docs/zuul-jobs/python-roles.html#rolevar-tox.tox_constraints_file\n\nChange-Id: I1a006cf4368cc18c34dd6f236ecfe966e6b09388\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/fc270afcc55df6471637fe4cd1f0295245bcf707"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/fc270afcc55df6471637fe4cd1f0295245bcf707"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
