)]}'
{"id":"openstack%2Foslo.limit~768303","triplet_id":"openstack%2Foslo.limit~master~Ic7a7b9cfb0607f652b86fa9098fe5ce26c74a4e9","project":"openstack/oslo.limit","branch":"master","topic":"ci","hashtags":[],"change_id":"Ic7a7b9cfb0607f652b86fa9098fe5ce26c74a4e9","subject":"[ci] fix openstack-lower-constraints-jobs","status":"ABANDONED","created":"2020-12-23 06:16:24.000000000","updated":"2021-01-05 05:14:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3ccc43cf01afe12ea277f2944080a11ec2123f52","_number":768303,"virtual_id_number":768303,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-23 07:46:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-23 07:46:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c4c473fe9c6670e08fd43d3fec85c0a79aba3c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-12-23 06:16:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86ab1267a619fcc7e6ce49523828a1d39d7b579d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 07:46:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6bd4e503a3a74f7d92415ac191f28617 : SUCCESS in 2m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1723319eea554e6fab0a4e0112481bab : SUCCESS in 1h 24m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35c6385252104a429c82ad6a43d69d13 : FAILURE in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5342c63a376344e6b50bfe50873e7da5 : FAILURE in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe7247e290f94d0bbdb090373f0e9a00 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2823dd8ec9b44c01aedbe0cfeb7c47b7 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ab9f5bac13d4f908d1e346ec41bc9b6 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ba3e085f72b45f7acb650e0ffa1cfb5 : SUCCESS in 4m 51s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/62344c63bda249ea9bd318a74947be8b : SUCCESS in 41m 55s","accounts_in_message":[],"_revision_number":1},{"id":"3ccc43cf01afe12ea277f2944080a11ec2123f52","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-01-05 05:14:58.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fdf4a168b36d1a3dc63b6bc2da34688427001611","revisions":{"fdf4a168b36d1a3dc63b6bc2da34688427001611":{"kind":"REWORK","_number":1,"created":"2020-12-23 06:16:24.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/03/768303/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/03/768303/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/03/768303/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/03/768303/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/03/768303/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/03/768303/1"}}},"commit":{"parents":[{"commit":"a5c678ce8b51ec76eefaa66f21752504ff42d2a4","subject":"Use py3 as the default runtime for tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/a5c678ce8b51ec76eefaa66f21752504ff42d2a4"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-12-23 06:04:50.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-12-23 06:15:35.000000000","tz":480},"subject":"[ci] fix openstack-lower-constraints-jobs","message":"[ci] fix openstack-lower-constraints-jobs\n\nChange-Id: Ic7a7b9cfb0607f652b86fa9098fe5ce26c74a4e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/fdf4a168b36d1a3dc63b6bc2da34688427001611"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/fdf4a168b36d1a3dc63b6bc2da34688427001611"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
