)]}'
{"id":"openstack%2Foslo.limit~779955","triplet_id":"openstack%2Foslo.limit~master~I504c93887413d5e19c5ffeb766dc8d047a5d357f","project":"openstack/oslo.limit","branch":"master","topic":"fix_hacking_version_pre_commit","hashtags":[],"change_id":"I504c93887413d5e19c5ffeb766dc8d047a5d357f","subject":"Fix the conflict status with hacking.","status":"ABANDONED","created":"2021-03-11 10:11:12.000000000","updated":"2021-03-22 13:45:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dafefa40891cab10669b8771479aea4fba10d29b","_number":779955,"virtual_id_number":779955,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-18 14:08:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-11 10:11:45.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2021-03-11 10:11:45.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-11 11:24:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b25f8b7874423e5a06f31f1e3d4e50f6473df579","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-03-11 10:11:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41c492286ebef4ad905f1d974055f32729a91082","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 11:24:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/12518ec64aa446e986443b4329a3123d : SUCCESS in 1h 08m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de63d69bdde84472b9c6d871e5003be5 : FAILURE in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bbff54aa55a4deb9b78452fb9087b96 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efcebafa8a21441e99951ff26027e320 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2dc59bdb163b425e815981cd28e72646 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/237d638b7a6942aa91b5257566d89a50 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/167c53cb5b484ca2835310e1c9749890 : SUCCESS in 4m 26s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/2e9ba0cf936544c88c6a0fe71b5c2e22 : SUCCESS in 31m 09s","accounts_in_message":[],"_revision_number":1},{"id":"7bcff122b95c3265204a74847dc1df2d26250676","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-03-18 11:28:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"62ca55a8df2c461326cedabb5c5050cadaf577d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 14:08:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c5ca77d2a6a6401191f39eea1683120f : FAILURE in 4m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/423d00690ec740b8b146adbb7739c703 : SUCCESS in 1h 30m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/988d1d35837c49048deb6398ef39e622 : SUCCESS in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42194b1adcb2490aa9ac396cfbbb2f9f : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2af9811ecb8441e8efe67b901675357 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64e54f87d5a0454eb9ff757b694a803c : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f50c415d1ea4ba7a3f67d99b9d2a8e6 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4570877eb3e54bb3bf040965ab8892cb : SUCCESS in 6m 38s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/5124cf851ae44da5abc69be29f717f76 : SUCCESS in 29m 34s","accounts_in_message":[],"_revision_number":2},{"id":"dafefa40891cab10669b8771479aea4fba10d29b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-03-22 13:45:18.000000000","message":"Abandoned\n\nUse the new approach[1] flake8 as local.\n\n[1] https://review.opendev.org/c/openstack/oslo.limit/+/782257","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"01aa8825c362d2f6cc4b988463cffa2e3bcdd046","revisions":{"dd8251b01cf7f7d5e31763902af80a18d54acbbc":{"kind":"REWORK","_number":1,"created":"2021-03-11 10:11:12.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/55/779955/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/55/779955/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/55/779955/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/55/779955/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/55/779955/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/55/779955/1"}}},"commit":{"parents":[{"commit":"a5c678ce8b51ec76eefaa66f21752504ff42d2a4","subject":"Use py3 as the default runtime for tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/a5c678ce8b51ec76eefaa66f21752504ff42d2a4"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-03-11 10:17:46.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-03-11 10:17:46.000000000","tz":60},"subject":"Fix the conflict status with hacking.","message":"Fix the conflict status with hacking.\n\nThe conflict is caused by:\n    The user requested flake8 3.8.3 (from /home/zuul/.cache/pre-commit/repozi7hytk7)\n    hacking 3.0.1 depends on flake8\u003c3.8.0 and \u003e\u003d3.6.0\n\nInstall the latest version of hacking to fix the conflict error with the\npre-commit hook.\n\nChange-Id: I504c93887413d5e19c5ffeb766dc8d047a5d357f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/dd8251b01cf7f7d5e31763902af80a18d54acbbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/dd8251b01cf7f7d5e31763902af80a18d54acbbc"}]},"branch":"refs/heads/master"},"01aa8825c362d2f6cc4b988463cffa2e3bcdd046":{"kind":"REWORK","_number":2,"created":"2021-03-18 11:28:28.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/55/779955/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/55/779955/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/55/779955/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/55/779955/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/55/779955/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/55/779955/2"}}},"commit":{"parents":[{"commit":"a5c678ce8b51ec76eefaa66f21752504ff42d2a4","subject":"Use py3 as the default runtime for tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/a5c678ce8b51ec76eefaa66f21752504ff42d2a4"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-03-11 10:17:46.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-03-18 11:34:21.000000000","tz":60},"subject":"Fix the conflict status with hacking.","message":"Fix the conflict status with hacking.\n\nThe conflict is caused by:\n    The user requested flake8 3.8.3 (from /home/zuul/.cache/pre-commit/repozi7hytk7)\n    hacking 3.0.1 depends on flake8\u003c3.8.0 and \u003e\u003d3.6.0\n\nInstall the latest version of hacking to fix the conflict error with the\npre-commit hook.\n\nFix the bandit version.\nTo fix:\n\nERROR: Could not find a version that satisfies the requirement bandit\u003c1.7.0,\u003e\u003d1.6.0\nERROR: No matching distribution found for bandit\u003c1.7.0,\u003e\u003d1.6.0\n\nFix the version of bandit in the lower-constraints.txt file to match\nwith the version in the test-requirements.txt file.\n\nChange-Id: I504c93887413d5e19c5ffeb766dc8d047a5d357f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/01aa8825c362d2f6cc4b988463cffa2e3bcdd046"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/01aa8825c362d2f6cc4b988463cffa2e3bcdd046"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
