)]}'
{"id":"openstack%2Foslo.limit~781368","triplet_id":"openstack%2Foslo.limit~stable%2Fwallaby~I93babc9a228d91c5d94b96e411c748326ad227d2","project":"openstack/oslo.limit","branch":"stable/wallaby","topic":"create-wallaby","hashtags":[],"change_id":"I93babc9a228d91c5d94b96e411c748326ad227d2","subject":"Update .gitreview for stable/wallaby","status":"MERGED","created":"2021-03-18 11:18:46.000000000","updated":"2021-04-16 12:45:48.000000000","submitted":"2021-04-16 12:43:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781368-create-wallaby","meta_rev_id":"011f35ca1b9eaf428c21815a8285528615684c78","_number":781368,"virtual_id_number":781368,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-16 12:43:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-04-15 14:35:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-04-15 14:35:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-18 12:08:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-15 14:25:37.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"fb97490cca957acc0a3639d54b685b7f586f774d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-03-18 11:18:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b720935416af4fd7cb3b509409b23250434e10c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 12:08:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c01d72df16448c88f458ab0d85478b8 : FAILURE in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1465a10deb34a6eba7988bdb3bc9015 : FAILURE in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b858dbc236d4995b861222f710c5a49 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd03f7237efb42bb8ffbf9b057e6f794 : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c0837ecef0b4210bf9d2aba30510c07 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bc128650e2b441ebc80eb58fdf4513b : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"02c9e4b4242097ea5854cf7de0e624295814f481","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-15 14:25:37.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"00f5692977eb31441f1c658257cd4319136fb98b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-15 14:35:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3bbedae1b15f0d380a6a37fcdc08a6ed0e6794ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 15:38:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/08fb428def914fb6b3678bb25b3ca06c : FAILURE in 1h 09m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce2b20a817f64564bd3e534475bc2fba : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ba12b1ee49d45eb90a3a725cfd6e025 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e0007bb7d55452e92a97f227b382f58 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75094b32745b4513b71a18df7e0ceb52 : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cfea628ae4d429fb5ee0268002997a4 : SUCCESS in 4m 00s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/e30adffd10c64b1099da3211fdfca37c : SUCCESS in 36m 16s","accounts_in_message":[],"_revision_number":2},{"id":"1f30f31861ce3591cee2282911b0bea4b233e70f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-16 09:33:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3a6d14a34767913cbd0820d7f14e2b031ccb011a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 11:00:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fd4dcde285bb44b482e1209c035b8f7c : SUCCESS in 1h 25m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61e29a32fdcc4e578c5ea6bea2b69d4e : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e553f9c289e4d38b1a6d90c06325213 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eed9cf2508494aa3ba0837be10b6eb60 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b577126a1d3042d0b3405c9190f09818 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cd36126e9a84a03bbbb7fff98f77f24 : SUCCESS in 4m 03s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/14ffb9b8e2fe4215a11d1a128660f135 : SUCCESS in 26m 54s","accounts_in_message":[],"_revision_number":2},{"id":"94c9cc9d9dd9b4d27a7f5e4f6ffea7f4d36cec1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 11:00:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a1c86bf5d417395fa3d8a83c08690b4b63724a00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 12:43:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e52b9ff6dea44d4f8314ce0decf6dd74 : SUCCESS in 1h 37m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff04798cb11f4de4b1b7b2271ca19b94 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1a182440fd947709c515948130afd7d : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0aab1803a534297aded54214b59eb24 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d9bd0cb99ca4b4da41116599681c492 : SUCCESS in 4m 01s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/d6e054c279894462825adf94a611023b : SUCCESS in 34m 00s","accounts_in_message":[],"_revision_number":2},{"id":"a3aa11e393309a82a200a4c49f8203b8be02ef3a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 12:43:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"011f35ca1b9eaf428c21815a8285528615684c78","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 12:45:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a9c76114bfd49afb197cdcfdbff273b : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2234552a41dce084f7ffbbf42e19f51cf3f138f2","revisions":{"29db18d22812d83453a14af8f686ba06225526a7":{"kind":"REWORK","_number":1,"created":"2021-03-18 11:18:46.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/68/781368/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/68/781368/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/68/781368/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/68/781368/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/68/781368/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/68/781368/1"}}},"commit":{"parents":[{"commit":"a5c678ce8b51ec76eefaa66f21752504ff42d2a4","subject":"Use py3 as the default runtime for tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/a5c678ce8b51ec76eefaa66f21752504ff42d2a4"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-18 11:18:45.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-18 11:18:45.000000000","tz":0},"subject":"Update .gitreview for stable/wallaby","message":"Update .gitreview for stable/wallaby\n\nChange-Id: I93babc9a228d91c5d94b96e411c748326ad227d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/29db18d22812d83453a14af8f686ba06225526a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/29db18d22812d83453a14af8f686ba06225526a7"}]},"branch":"refs/heads/stable/wallaby"},"2234552a41dce084f7ffbbf42e19f51cf3f138f2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-04-15 14:25:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/68/781368/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/68/781368/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/68/781368/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/68/781368/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/68/781368/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/68/781368/2"}}},"commit":{"parents":[{"commit":"2c63f8f2b67273f70c3f98efbd9038ce10a8c5f9","subject":"Fix requirements issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/2c63f8f2b67273f70c3f98efbd9038ce10a8c5f9"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-18 11:18:45.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-04-15 14:25:21.000000000","tz":60},"subject":"Update .gitreview for stable/wallaby","message":"Update .gitreview for stable/wallaby\n\nChange-Id: I93babc9a228d91c5d94b96e411c748326ad227d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/2234552a41dce084f7ffbbf42e19f51cf3f138f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/2234552a41dce084f7ffbbf42e19f51cf3f138f2"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
