)]}'
{"id":"openstack%2Foslo.limit~812096","triplet_id":"openstack%2Foslo.limit~master~I22234e0bb6b3a1cecb29a6b99a3afcd02ffdbf5f","project":"openstack/oslo.limit","branch":"master","topic":"bp/unified-limits-nova","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2022-01-24 17:28:49.000000000","reason":"melwitt replied on the change"}},"hashtags":[],"change_id":"I22234e0bb6b3a1cecb29a6b99a3afcd02ffdbf5f","subject":"Add interfaces for getting limits without enforcing","status":"MERGED","created":"2021-10-01 01:17:15.000000000","updated":"2022-01-24 19:04:00.000000000","submitted":"2022-01-24 19:01:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812096-bp/unified-limits-nova","meta_rev_id":"f935bac66fe095e15eb93934a5d60551ab92aaa3","_number":812096,"virtual_id_number":812096,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-24 19:01:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-01-24 17:21:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2022-01-24 16:56:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-01-24 17:21:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-01 02:40:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-24 16:56:28.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-01-24 17:21:32.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"2ac74aa361e5316d527c491c4ab246a91a864e4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-10-01 01:17:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"740257c1ccb1b5065d684f14ba281e6145c7e9c2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-10-01 01:17:48.000000000","message":"Topic fix-bug removed","accounts_in_message":[],"_revision_number":1},{"id":"6c44d532d3f2ef7e248bc20262bfdbc97967853f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-10-01 01:17:55.000000000","message":"Topic set to bp/unified-limits-nova","accounts_in_message":[],"_revision_number":1},{"id":"58347ac750d1ee193d7913f8c4efa545b291988e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 02:40:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a52314e8c2b3420196755c7feee5838f : SUCCESS in 1h 19m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb2fd66e07e94fd295d94c980125315c : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ed7693b484346a991c62d1f809fbe51 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9c9c2a465434f6fb173bbeaf9d36b83 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2894aff37c642c39a8b9a97dfb39098 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f5a2c988f8248b6bc375f8a6231144d : SUCCESS in 4m 07s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/364c5305f9b74e65ac2b002c48765929 : SUCCESS in 23m 55s","accounts_in_message":[],"_revision_number":1},{"id":"c96779c34dcfcccf9aff1933a5e00e682c7f3069","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-10-02 01:01:46.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"9ac637311385dc82e3ea8a1074bfa21fff8c2b0d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-10-02 01:02:05.000000000","message":"Topic fix-bug removed","accounts_in_message":[],"_revision_number":2},{"id":"bf0072e858424cad56d7e9e1e5e33f939af30392","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-10-02 01:02:13.000000000","message":"Topic set to bp/unified-limits-nova","accounts_in_message":[],"_revision_number":2},{"id":"f5545b0c706810d6ec9a52acb0c5ce6d43921ba3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-02 02:21:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7c17ba932ba54915a3b2fcafddbcf59d : SUCCESS in 1h 17m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22a3ac3e5ba149a2a3533d475d916868 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3de8e7f12deb49e29d7fd2018c8dea8f : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4a8d1b0829a4e9e913d7eeb16f18832 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3808aeea173243dcbdd86366ba65054b : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7a572f3c78b4151b0fdf196532db77d : SUCCESS in 5m 59s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/904b2e53f1654b5abc6552d572b1a622 : SUCCESS in 27m 48s","accounts_in_message":[],"_revision_number":2},{"id":"bb36c6907842486f84fac3e3b8a1b9b5fc4e0b37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2022-01-21 00:44:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"99097dabab0b48129de793757706031090b05d6e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2022-01-21 00:44:31.000000000","message":"Topic fix-bug removed","accounts_in_message":[],"_revision_number":3},{"id":"bb1cf75cdbd513d0230a78133a3b1b399bdfce01","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2022-01-21 00:44:35.000000000","message":"Topic set to bp/unified-limits-nova","accounts_in_message":[],"_revision_number":3},{"id":"4f7a8eb62ddfba8a194ed632fe73b1d1dc0862c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 02:20:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e06f0f0e69bc44d79fbe374b9c550af2 : SUCCESS in 1h 30m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e909bfd839744a90adac7123371cf68a : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e1bc06533fb4cd495fc7123bb0b47ab : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ae52e34344141a6b23965f2be398666 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98b7382ec2644fcdab4f9589191db61b : SUCCESS in 4m 08s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/1b0473e5b7d04b7389b87278e9e00335 : SUCCESS in 34m 11s","accounts_in_message":[],"_revision_number":3},{"id":"aa23de7eae0958450583bd9a863f360f61a3c5d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2022-01-21 19:35:46.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent, and commit message as Patch Set 3.","accounts_in_message":[],"_revision_number":4},{"id":"1ff799f2df884a735cc907f5f1f707cbb368176d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2022-01-21 19:48:33.000000000","message":"Topic fix-bug removed","accounts_in_message":[],"_revision_number":4},{"id":"04d1942f507292f644ed54e93eec1a0502a626c2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2022-01-21 19:48:47.000000000","message":"Topic set to bp/unified-limits-nova","accounts_in_message":[],"_revision_number":4},{"id":"696332b4cb2dc7134a06590d443dee7af6bb67a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 21:14:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02261226e24d4600935e330851eedfff : SUCCESS in 1h 33m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/341611a4c70c45e2812bb18ee223442b : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75300d15de624ac39b92a6474bc41ce1 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df60524fb08a4ab986e494305b13924d : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2c257f03dd6405dbea3014f6b8976cb : SUCCESS in 3m 43s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/1bc575d5378d4a3d8360837277313623 : SUCCESS in 24m 11s","accounts_in_message":[],"_revision_number":4},{"id":"a9bd5d3dabd0d2975358aca3113ac1aecf0e21a9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-01-24 16:56:28.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"961a848e69470cf8616c04e32e098d4bd60c41b1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-24 17:21:32.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"52283ae7aca0387de4d7700d47f66113c087aba3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 17:22:31.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"babb5b5b04a3bf4b7ff7b0b59122571e7fd63fdc","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2022-01-24 17:28:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"61640d536190fe5fbae464b8648d460ed9cfa26d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 19:01:40.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5585cb21a39c40d589489601f8bdd3f4 : SUCCESS in 1h 34m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5f691d4c3b94c339cb927f2f3059c44 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2af541dcd9ba4a709e14118cc9525ef6 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d760c3d3e4de4887a5f7f7fd580300f6 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c30507034db8412798d3da5fae3936ee : SUCCESS in 4m 52s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/612be6f4f1e34d97b1713f9b56ea8306 : SUCCESS in 26m 42s","accounts_in_message":[],"_revision_number":4},{"id":"e96b16eb94b9e5e25e86bc5f482d0206d071a8b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 19:01:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"f935bac66fe095e15eb93934a5d60551ab92aaa3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 19:04:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb8cc3a8a56943b3b86fa4c54b922a8b : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bf9deb10c3af4c43f6952f0fcedb9d48dfd8340d","revisions":{"5bc41a004ed107bcb0e54e5050f11b73c7501916":{"kind":"REWORK","_number":1,"created":"2021-10-01 01:17:15.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/96/812096/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/96/812096/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/96/812096/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/96/812096/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/96/812096/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/96/812096/1"}}},"commit":{"parents":[{"commit":"e9c038ef3bb035909fbd421eb7a679fad410ca04","subject":"Make calculate_usage() work if limits are missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/e9c038ef3bb035909fbd421eb7a679fad410ca04"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-10-01 01:03:43.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-10-01 01:09:52.000000000","tz":0},"subject":"WIP Add interfaces for getting limits without enforcing","message":"WIP Add interfaces for getting limits without enforcing\n\nWe currently have a public Enforcer interface for getting limits and\ncalculating usage but it is not yet possible to retrieve only limits\nwithout calculating usage.\n\nWhile working on unified limits support in nova, we realized we need a\nway to get limits only. In nova there are legacy APIs for showing quota\nlimits and initially we will provide compat by proxying to keystone to\nget the limits.\n\nThis adds public interfaces for getting limits to Enforcer.\n\nRelated to blueprint unified-limits-nova\n\nChange-Id: I22234e0bb6b3a1cecb29a6b99a3afcd02ffdbf5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/5bc41a004ed107bcb0e54e5050f11b73c7501916"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/5bc41a004ed107bcb0e54e5050f11b73c7501916"}]},"branch":"refs/heads/master"},"47acb648ddb9bc6169984e54f2f03b13082de758":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-10-02 01:01:46.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/96/812096/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/96/812096/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/96/812096/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/96/812096/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/96/812096/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/96/812096/2"}}},"commit":{"parents":[{"commit":"d1b92e557c29be7afd680f49e626650ad4485dad","subject":"WIP: Allow project_id\u003dNone for enforce/calculate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/d1b92e557c29be7afd680f49e626650ad4485dad"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-10-01 01:03:43.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-10-02 01:01:29.000000000","tz":0},"subject":"WIP Add interfaces for getting limits without enforcing","message":"WIP Add interfaces for getting limits without enforcing\n\nWe currently have a public Enforcer interface for getting limits and\ncalculating usage but it is not yet possible to retrieve only limits\nwithout calculating usage.\n\nWhile working on unified limits support in nova, we realized we need a\nway to get limits only. In nova there are legacy APIs for showing quota\nlimits and initially we will provide compat by proxying to keystone to\nget the limits.\n\nThis adds public interfaces for getting limits to Enforcer.\n\nRelated to blueprint unified-limits-nova\n\nChange-Id: I22234e0bb6b3a1cecb29a6b99a3afcd02ffdbf5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/47acb648ddb9bc6169984e54f2f03b13082de758"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/47acb648ddb9bc6169984e54f2f03b13082de758"}]},"branch":"refs/heads/master"},"97f4cfe5737cab2f762796d0a8b7c15ff4c3655b":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2022-01-21 00:44:01.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/96/812096/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/96/812096/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/96/812096/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/96/812096/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/96/812096/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/96/812096/3"}}},"commit":{"parents":[{"commit":"1dd3ce81bf678e6bf687f9fd33173771b47dd5ff","subject":"WIP: Allow project_id\u003dNone for enforce/calculate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/1dd3ce81bf678e6bf687f9fd33173771b47dd5ff"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-10-01 01:03:43.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2022-01-21 00:39:45.000000000","tz":0},"subject":"Add interfaces for getting limits without enforcing","message":"Add interfaces for getting limits without enforcing\n\nWe currently have a public Enforcer interface for getting limits and\ncalculating usage but it is not yet possible to retrieve only limits\nwithout calculating usage.\n\nWhile working on unified limits support in nova, we realized we need a\nway to get limits only. In nova there are legacy APIs for showing quota\nlimits and initially we will provide compat by proxying to keystone to\nget the limits.\n\nThis adds public interfaces for getting limits to Enforcer.\n\nRelated to blueprint unified-limits-nova\n\nChange-Id: I22234e0bb6b3a1cecb29a6b99a3afcd02ffdbf5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/97f4cfe5737cab2f762796d0a8b7c15ff4c3655b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/97f4cfe5737cab2f762796d0a8b7c15ff4c3655b"}]},"branch":"refs/heads/master"},"bf9deb10c3af4c43f6952f0fcedb9d48dfd8340d":{"kind":"NO_CHANGE","_number":4,"created":"2022-01-21 19:35:46.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/96/812096/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/96/812096/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/96/812096/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/96/812096/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/96/812096/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/96/812096/4"}}},"commit":{"parents":[{"commit":"7e4f36abdb72cea991b1b99a67d30b088cc60d7d","subject":"Allow project_id\u003dNone for enforce/calculate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/7e4f36abdb72cea991b1b99a67d30b088cc60d7d"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-10-01 01:03:43.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2022-01-21 19:35:33.000000000","tz":0},"subject":"Add interfaces for getting limits without enforcing","message":"Add interfaces for getting limits without enforcing\n\nWe currently have a public Enforcer interface for getting limits and\ncalculating usage but it is not yet possible to retrieve only limits\nwithout calculating usage.\n\nWhile working on unified limits support in nova, we realized we need a\nway to get limits only. In nova there are legacy APIs for showing quota\nlimits and initially we will provide compat by proxying to keystone to\nget the limits.\n\nThis adds public interfaces for getting limits to Enforcer.\n\nRelated to blueprint unified-limits-nova\n\nChange-Id: I22234e0bb6b3a1cecb29a6b99a3afcd02ffdbf5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/bf9deb10c3af4c43f6952f0fcedb9d48dfd8340d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/bf9deb10c3af4c43f6952f0fcedb9d48dfd8340d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
