)]}'
{"id":"openstack%2Foslo.limit~855895","triplet_id":"openstack%2Foslo.limit~stable%2Fussuri~I562dcde92227a8d5827de6fec4871e83365f9bea","project":"openstack/oslo.limit","branch":"stable/ussuri","topic":"cap-virtualenv-py37","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-09-19 18:23:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I562dcde92227a8d5827de6fec4871e83365f9bea","subject":"[stable-only] Cap virtualenv for py37","status":"MERGED","created":"2022-09-05 13:56:03.000000000","updated":"2022-09-19 18:27:39.000000000","submitted":"2022-09-19 18:23:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855895-cap-virtualenv-py37","meta_rev_id":"ce942c0b76a0484c3127addc50e037978d4327bc","_number":855895,"virtual_id_number":855895,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-19 18:23:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-09-14 18:08:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-09-19 16:50:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-05 15:46:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-14 16:19:38.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2022-09-14 18:08:44.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"addbc09ceedcf4263e849cb731904b640d919fb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-05 13:56:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"807516d626b0906dfba6379b7b72e9b79d58e012","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 15:46:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7bb9f89c3184192a08bfb14036869af\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ef49bba7f6f49348b28e88078610ea7 : SUCCESS in 1h 43m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f280129b5f3545b19deaf76a15fda937 : FAILURE in 2m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c14fd7581a01431da0d79b9ddb21914c : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d349ab1fac934f74a99efb5c72902617 : SUCCESS in 2m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b40b37ded1144e28338456564933d86 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/241625a7f96f4bac9cf7fa6a1933e3d3 : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87a35237bc184563a1e96be66d6ed306 : SUCCESS in 2m 48s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/b1ca1c83a84d440ca9bc2651b5092a7e : SUCCESS in 35m 18s","accounts_in_message":[],"_revision_number":1},{"id":"42f882ca3e3021652aa3053a4a4fc8f4aa70bb96","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-09-14 16:19:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2a03edd2951cc0d9ea0789977bbf2c510c44aa33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 17:42:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f404f854d86a4f019bd9cb0b0cd2321a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d141a4f389554321854a1a4137dfb2e8 : SUCCESS in 1h 21m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2eb48419cf844a690cee796415b3242 : FAILURE in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7264d8f5710a41c28f4717f7d4d58a6c : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed9bf206225d4700b2a4b08ee7fdb118 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/091698c55a974c5991d9e1d39c2df602 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/147999df680b49bea356b2251fb8b8eb : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/220716f9387a40cc8a000d7c9162cb76 : SUCCESS in 3m 40s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/003a65376cef40b3af31eac584575fae : SUCCESS in 28m 47s","accounts_in_message":[],"_revision_number":1},{"id":"59198f3b0f2434ec6ec1d4682c44e0b0b5fe9db4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-14 18:00:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b80d72be145166ecdfdabfb39c1c62c722b84c42","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-14 18:08:44.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bdc211a4ade9ccc51328f656360f74a7eefdfc53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 19:29:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/464596a415b8461fb8a808f4c4f22dcf\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/79d0c02df8b84648b8edaf8e70b87dc1 : SUCCESS in 1h 27m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0f602b2cd6c4ad7ac50933e52a27c15 : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1232d98c3274b9db332fe5496b4cee5 : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/58951d4dac6148e391f15c52e33c98be : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a59a81b0c133486dbc1f4ce8f633cc16 : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89d44ae44eb04d088f332fa49ff449fd : SUCCESS in 3m 10s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/b670fbe947364df49d8ad960a20c9195 : SUCCESS in 34m 33s","accounts_in_message":[],"_revision_number":2},{"id":"b585b6f2e41c556950933b620e0554f9a025f9cc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-19 16:50:03.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f03fcb63fd0d31b0d23b18b5eb65799d68b86257","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 16:51:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6f5a73d915de3c84d25fc0aeb2f30835e1329821","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 18:23:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48f85dd8327b41f98b444627585c1520\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4333fa6729a947de8cb386662226ef2f : SUCCESS in 1h 25m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9b9ac50dd7c438087a555d2181365ab : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8976ff343e214bc9b6e64a854a3db449 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7ec39811aeaf42ba83d3d620906a194e : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30c3a9cc9b0441da83bd44580eaea590 : SUCCESS in 4m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0652ed24dc62439e9494b7e74c9c5dd1 : SUCCESS in 3m 58s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/cf2d37cd926745948726847e7ef43780 : SUCCESS in 36m 18s","accounts_in_message":[],"_revision_number":2},{"id":"b2d743e4d576d1013e00f3b579bc0346469d784e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 18:23:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ce942c0b76a0484c3127addc50e037978d4327bc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 18:27:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5a592dfd32748c1ad635c9a946194d5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ff0bda19d7f4816ba9bfc60f34fbedc : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/341aa4896f024c5f93a434407e8127e5 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"80bbd9cde33eb39cc52c4b77c941aae9a8d44513","revisions":{"0a5ff8acaad5fc9bd2a2f65ec1adfeca3839c4ce":{"kind":"REWORK","_number":1,"created":"2022-09-05 13:56:03.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/95/855895/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/95/855895/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/95/855895/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/95/855895/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/95/855895/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/95/855895/1"}}},"commit":{"parents":[{"commit":"824d1d8250e3b2a64efcc40365aeeff9b43ab416","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/824d1d8250e3b2a64efcc40365aeeff9b43ab416"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-05 13:54:54.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-05 13:54:54.000000000","tz":120},"subject":"[stable-only] Cap virtualenv for py37","message":"[stable-only] Cap virtualenv for py37\n\npy37 test started to fail with latest virtualenv (20.16.4) as it\nbundles setuptools 65.3.0, which causes installation problems under\npy37. (Note: with py38 the problem does not occur\n\nThis patch sets virtualenv\u003c20.16.4 to unblock the gate as\nvirtualenv 20.16.3 bundles setuptools 63.4.1, with which py37 gate\nworks well.\n\nChange-Id: I562dcde92227a8d5827de6fec4871e83365f9bea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/0a5ff8acaad5fc9bd2a2f65ec1adfeca3839c4ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/0a5ff8acaad5fc9bd2a2f65ec1adfeca3839c4ce"}]},"branch":"refs/heads/stable/ussuri"},"80bbd9cde33eb39cc52c4b77c941aae9a8d44513":{"kind":"REWORK","_number":2,"created":"2022-09-14 18:00:30.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/95/855895/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/95/855895/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/95/855895/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/95/855895/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/95/855895/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/95/855895/2"}}},"commit":{"parents":[{"commit":"824d1d8250e3b2a64efcc40365aeeff9b43ab416","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/824d1d8250e3b2a64efcc40365aeeff9b43ab416"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-05 13:54:54.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-14 18:00:29.000000000","tz":60},"subject":"[stable-only] Cap virtualenv for py37","message":"[stable-only] Cap virtualenv for py37\n\npy37 test started to fail with latest virtualenv (20.16.4) as it\nbundles setuptools 65.3.0, which causes installation problems under\npy37. (Note: with py38 the problem does not occur\n\nThis patch sets virtualenv\u003c20.16.4 to unblock the gate as\nvirtualenv 20.16.3 bundles setuptools 63.4.1, with which py37 gate\nworks well.\n\nNOTE(stephenfin): Modified to remove the (failing) l-c job.\n\nChange-Id: I562dcde92227a8d5827de6fec4871e83365f9bea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/80bbd9cde33eb39cc52c4b77c941aae9a8d44513"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/80bbd9cde33eb39cc52c4b77c941aae9a8d44513"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
