)]}'
{"id":"openstack%2Foslo.limit~858740","triplet_id":"openstack%2Foslo.limit~stable%2Ftrain~I562dcde92227a8d5827de6fec4871e83365f9bea","project":"openstack/oslo.limit","branch":"stable/train","topic":"cap-virtualenv-py37","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-09-26 16:37:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I562dcde92227a8d5827de6fec4871e83365f9bea","subject":"[stable-only] Cap virtualenv for py37","status":"MERGED","created":"2022-09-21 14:33:27.000000000","updated":"2022-09-26 16:39:17.000000000","submitted":"2022-09-26 16:37:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858740-cap-virtualenv-py37","meta_rev_id":"8fb4a55bfa37684c03060dbf4ac8271c208d1119","_number":858740,"virtual_id_number":858740,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-26 16:37:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2022-09-26 14:46:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":1,"date":"2022-09-26 14:46:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-21 16:09:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-26 14:46:38.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"41511fa54a534f573677ee61e76cf871ac9d29cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-21 14:33:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e344efe6cb5fc3b60400d88da7f2735757903943","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 16:09:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f0824ec506a458e924ce60208f742bb\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/7e7fd734c9384fb298d263216468cd5f : SUCCESS in 1h 29m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f969d48b1b8e4a5b8da9066782ca4d68 : SUCCESS in 1h 33m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7561a6513f24dcd869e03f3b88e564e : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ee6ebfedea5246a0b7663317bb421767 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3c1a69eb0024be180d2606039ae8d49 : SUCCESS in 7m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e03c3a407a3d49479db5eb0b52bcdf2e : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/debec187405a42c2baa4ec98a8438853 : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41af55228cb74b6cab824a8852c4098d : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"4c6942484ee228ef8febaf5c11b3a3a14416a742","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-09-26 14:46:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"88cc1dc2f53542ca7ee024588db5fcbb63206ac4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 14:47:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5bf010c64a7e5f16236c2a1862a75cdc3057cac1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 16:37:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6ad59b848c64f2fbabeee6056dafae0\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/3a51b5c98a254b80a6e4c22d130ed607 : SUCCESS in 1h 45m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e84bd1dda0eb4914b1ed176650b50502 : SUCCESS in 1h 18m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69d614b5714d4caeab5331abbefb6296 : SUCCESS in 3m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f2b30b5579c84adcb1de38fb42019311 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03fe838b35274b6fa5f52fae650e9b9d : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9534eb8b2a4a44e08857f23432f47d15 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ac28fc21115413eb11fe70768cd7554 : SUCCESS in 4m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/838fbd87a0ab48a2829895d670273cc0 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"d5a8db88b2916c4878c7706405a1f4b525eb1606","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 16:37:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8fb4a55bfa37684c03060dbf4ac8271c208d1119","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 16:39:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cfcbf6d9e9348a88844a3ff90451694\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/063be2735f814d3fa5123e0e930ae5d0 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03434ffcf8ef435e95874a7496ed6730 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"51edcdd4a70ed4ad788ec98cb2f43fb57401ea13","revisions":{"51edcdd4a70ed4ad788ec98cb2f43fb57401ea13":{"kind":"REWORK","_number":1,"created":"2022-09-21 14:33:27.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/40/858740/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/40/858740/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/40/858740/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/40/858740/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/40/858740/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/40/858740/1"}}},"commit":{"parents":[{"commit":"45eeda832a8f65fdfdcd79b1b4c9eb30cee09b09","subject":"[stable-only] Cap bandit and fix lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/45eeda832a8f65fdfdcd79b1b4c9eb30cee09b09"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-05 13:54:54.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-21 14:33:06.000000000","tz":120},"subject":"[stable-only] Cap virtualenv for py37","message":"[stable-only] Cap virtualenv for py37\n\npy37 test started to fail with latest virtualenv (20.16.4) as it\nbundles setuptools 65.3.0, which causes installation problems under\npy37. (Note: with py38 the problem does not occur\n\nThis patch sets virtualenv\u003c20.16.4 to unblock the gate as\nvirtualenv 20.16.3 bundles setuptools 63.4.1, with which py37 gate\nworks well.\n\nNOTE(stephenfin): Modified to remove the (failing) l-c job.\n\nConflicts:\n  .zuul.yaml\n  tox.ini\n\nNOTE(elod.illes): conflict in .zuul.yaml is caused by ussuri job\n  template changes; conflict in tox.ini is caused by tox.ini cleanup\n  in ussuri: I197d6486a3b126f616ffe2f7e3e669d922a1e337\n\nChange-Id: I562dcde92227a8d5827de6fec4871e83365f9bea\n(cherry picked from commit 80bbd9cde33eb39cc52c4b77c941aae9a8d44513)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/51edcdd4a70ed4ad788ec98cb2f43fb57401ea13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/51edcdd4a70ed4ad788ec98cb2f43fb57401ea13"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
