)]}'
{"id":"openstack%2Foslo.limit~873992","triplet_id":"openstack%2Foslo.limit~master~I01903ee431adb47901cf9927f369178ab20c9ca4","project":"openstack/oslo.limit","branch":"master","topic":"release-health-check-cwi-antelope","hashtags":[],"change_id":"I01903ee431adb47901cf9927f369178ab20c9ca4","subject":"DNM: dummy change to test gate health","status":"ABANDONED","created":"2023-02-15 19:32:02.000000000","updated":"2023-02-16 14:31:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0b07e69621dffd2bb143a74f0f0e42bad6590453","_number":873992,"virtual_id_number":873992,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-02-15 20:55:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-15 20:55:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8211feafc7a95c77d7c2e7f29c65be59c284aa2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-15 19:32:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80a5a7b902b72c1105792e5faab4eea5593f1e90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 20:55:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76fff16429524dc68a1cd0054d97453c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef5f818a0a924c65887c3d63e462f02c : SUCCESS in 1h 22m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43f1472b722549efb35a0b6cd2942144 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67e36d9cefd14e83a271e95547363356 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b5102d4a1a34085ad937ffe2c7a64f6 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/587df9982ae24320af28dfe1ce924c2a : SUCCESS in 3m 44s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/5849a2c7c5524f9282a7d496cd8dd68c : SUCCESS in 21m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0b07e69621dffd2bb143a74f0f0e42bad6590453","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-16 14:31:49.000000000","message":"Abandoned\n\nGate looks healthy, patch can be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e680a5188814377a73f009814c880ecaf8345bdc","revisions":{"e680a5188814377a73f009814c880ecaf8345bdc":{"kind":"REWORK","_number":1,"created":"2023-02-15 19:32:02.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/92/873992/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/92/873992/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/92/873992/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/92/873992/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/92/873992/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/92/873992/1"}}},"commit":{"parents":[{"commit":"ee1d1aadd0802d763af95fc608621a675f8b0103","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/ee1d1aadd0802d763af95fc608621a675f8b0103"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-02-15 19:31:52.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-02-15 19:32:00.000000000","tz":60},"subject":"DNM: dummy change to test gate health","message":"DNM: dummy change to test gate health\n\nChange-Id: I01903ee431adb47901cf9927f369178ab20c9ca4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/e680a5188814377a73f009814c880ecaf8345bdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/e680a5188814377a73f009814c880ecaf8345bdc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
