)]}'
{"id":"openstack%2Foslo.limit~875148","triplet_id":"openstack%2Foslo.limit~stable%2F2023.1~Ic1c1d22ad2dc56c6a3e644e6552d85cad4bf6859","project":"openstack/oslo.limit","branch":"stable/2023.1","topic":"create-2023.1","hashtags":[],"change_id":"Ic1c1d22ad2dc56c6a3e644e6552d85cad4bf6859","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.1","status":"MERGED","created":"2023-02-24 15:18:42.000000000","updated":"2023-02-27 15:11:23.000000000","submitted":"2023-02-27 15:10:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875148-create-2023.1","meta_rev_id":"afcaf3a4203604ea76ac6d8ff03981dc21c4bb34","_number":875148,"virtual_id_number":875148,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-27 15:10:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2023-02-27 12:06:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2023-02-27 14:53:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2023-02-27 12:06:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-24 15:35:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-27 12:06:24.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-02-27 14:53:29.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"57ef6b154701834191f1a5ca64fc6ba9ff59e443","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2023-02-24 15:18:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6cfce1570f6db8319115e933fd177b9f9d4ff9cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 15:35:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90c47f33e9c24e7d82cc0be51b87aa55\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32a60f3f2be94a16bc8b24a7cf10be4e : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23c9fd5af00b481dbaab4517e38a244d : SUCCESS in 2m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d8c02d36f4eb4f73aba8a4ed53116622 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f369f0fdecc243e4b919c35b5f5c2c0a : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f3fa6d509a14fc8b88a5f2dc922be1d : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c179d009fba128e709fc8738d3369f5ac658656b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-27 12:06:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"642c088e89f9ffd8307886ef7857069a8e9bad19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 12:08:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cf93b51d408b6ea28ae3085ebafd895412920ed1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 12:24:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/68a13bd659054f4db486c06a73217a01\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1f4a7025e184cd1b7f50ebcee9dc74e : RETRY_LIMIT in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90afd06bbd064e4793c02244450072d8 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a784b872319b42ad9072f69daa05955d : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/089bfe1279b74adb852284f519c42d83 : SUCCESS in 2m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1cf2ea0dc67402d9e8c139f6a31a54c : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":1},{"id":"b355f7523148e0f2e18d26973ccfa824a2a8a446","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-27 14:53:29.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ad47e78b2163fc8bc71c3a4a0b3831ac25255b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 15:02:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/650bf2dd6bb94298b03f3906c2ab0d2f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/234c0b2558db46ffa1c72e961bb8ef3e : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b6045e8119144e7a5fe8d1112da43cf : SUCCESS in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe0e2adf2523455692cfde21c31fbac8 : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56792ffac4c7462e9e9b91f244b9e51a : SUCCESS in 3m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9eb41e4860954a37b462625d212a508b : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":1},{"id":"95d567e1d671272a1f8d3fd5a80ee76dc8815a11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 15:03:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4c5fe385ef54538327c6600ce073257d5d1a5395","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 15:10:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7976221556094226a23a252f46111197\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6aa3a5b918147beb4eaa00affc922ae : SUCCESS in 2m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a826e0806aa44a3cb6c5e9d6ef6bf01d : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/205eeacd3088441589e6c32f53ca498f : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7bc130635ae458c911711c0cbeb45c8 : SUCCESS in 3m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce1afbedcd1949109403f7ce3703eed5 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":1},{"id":"acbe0313286c039462b51af9fb4ab82a4b186a62","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 15:10:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"afcaf3a4203604ea76ac6d8ff03981dc21c4bb34","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 15:11:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb737433c3604e91b2a9642396a0359d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/133720d237914105abee5b2e8bea7a4b : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a48e7a43bffd45c2b1910182a86daa5f : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d0a78df30c30a08efb6c25c9a96c67f50993e216","revisions":{"d0a78df30c30a08efb6c25c9a96c67f50993e216":{"kind":"REWORK","_number":1,"created":"2023-02-24 15:18:42.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/48/875148/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/48/875148/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/48/875148/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/48/875148/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/48/875148/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/48/875148/1"}}},"commit":{"parents":[{"commit":"6fd104cff5b4ea5f17008bd19a166f36ad92ace6","subject":"Update .gitreview for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/6fd104cff5b4ea5f17008bd19a166f36ad92ace6"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-02-24 15:18:39.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-02-24 15:18:39.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.1","message":"Update TOX_CONSTRAINTS_FILE for stable/2023.1\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2023.1 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Ic1c1d22ad2dc56c6a3e644e6552d85cad4bf6859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/d0a78df30c30a08efb6c25c9a96c67f50993e216"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/d0a78df30c30a08efb6c25c9a96c67f50993e216"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
