)]}'
{"id":"openstack%2Foslo.limit~879283","triplet_id":"openstack%2Foslo.limit~master~I47b536149e8119324b3b715b2103b803b74a8a56","project":"openstack/oslo.limit","branch":"master","topic":"oslo-drop-py38","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-04-11 11:29:59.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-04-11 11:29:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I47b536149e8119324b3b715b2103b803b74a8a56","subject":"Moves supported python runtimes from version 3.8 to 3.10","status":"MERGED","created":"2023-04-03 09:06:29.000000000","updated":"2023-05-03 20:02:15.000000000","submitted":"2023-04-11 11:29:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879283-oslo-drop-py38","meta_rev_id":"1fd201658ba0f926822afc11bca5e1e4c165fd1b","_number":879283,"virtual_id_number":879283,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-11 11:29:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2023-04-11 09:57:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2023-04-11 09:57:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-03 10:33:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-10 01:46:31.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-04-11 09:57:37.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"e4620d6f06bfb839f18a4aae73cb5eaf27098c90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-04-03 09:06:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2afc6d39cb8c0ea1d2ea6b22305045425c11b58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 10:33:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e548a699e10b4a25ba95578befec6bdd\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ba1234cea142412aaeb42c03219c8c2b : SUCCESS in 1m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2f72c84569364a9aa0a629249246cdc3 : SUCCESS in 1h 19m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0278a70c6049435687e75f1e87e75c3e : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2323e17753f4308bb261c9914fd80d9 : SUCCESS in 2m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72d41d64608b46e29f50310c1598906a : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c645bca3f3124f9385e2228b06607a64 : SUCCESS in 3m 37s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/2e2bcbc8d9674b758cfd5d55d6de2f6c : SUCCESS in 28m 33s","accounts_in_message":[],"_revision_number":1},{"id":"ed1c2427e86de609275b92024db7e1337bea8eb4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-10 01:46:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"330c1efb30f623170762cfb046e06b2ac3848f3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-10 01:54:02.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b4fa1a5ef389805fe0c2a0c6412e0847ffbdd773","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-10 01:54:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6dd4341a76727e049dea0609f31b8e8d0d054d45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 03:21:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/878becb4980c43c897512921237dccb4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/baf6ed00696943a4a646d0f0036899da : SUCCESS in 1m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e218f4dd6e2a434cbf3e46f4b3d61222 : SUCCESS in 1h 17m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3467c2027b3417fbce013d164ea48cc : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/216199d4ce5944009ad3c1ad606f4a4a : FAILURE in 2m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/56f7991399cf48b69da28cfc3bcee733 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d921ab42f5a94b0dbdb5eb5ab28d287e : SUCCESS in 3m 31s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/f173281c037b4b61879ab867266fdf53 : SUCCESS in 20m 33s","accounts_in_message":[],"_revision_number":2},{"id":"ab18e70e6ffde65432e4120fe3033669d7a00b0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-10 04:06:22.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d316759606aad8be98edd02916a565b234f7fa58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 06:05:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9223ccd3d7442eb8b9656aa33e26217\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/73649746464449109168dd61bd7db430 : SUCCESS in 1m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7ee082c9c6904d3f977dcdf597736d76 : SUCCESS in 1h 51m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/578aef9910e54941bb25255dcebc7dc0 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/237d5832320d4cf28c08e61c0f70c330 : SUCCESS in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b5b57bf4dc14834917e0a904d773894 : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a1ae467ea894516b6a22323a65685f6 : SUCCESS in 3m 03s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/21058a813ab94edea1d15253ee832cba : SUCCESS in 27m 51s","accounts_in_message":[],"_revision_number":3},{"id":"1af104eca18cdde6bb505ef4a78d315fabcadc4a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-11 09:57:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cf99199da5cd9b4a8d9b2e2bd9386e42b8c2c562","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 09:58:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"43ed54ce2e58566514fe415428db073e189e9639","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 11:29:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c1873a90afb48f1af4479d2a99bf674\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ec8e1ce5d0e34384b76dfa771bc5df0c : SUCCESS in 1m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e01345aaaa00468d929b2a617a0c3b42 : SUCCESS in 1h 26m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5e59863185847a181f48ae64ca3dfc0 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1002f2ff9c5145eb82a82429c8cb13a3 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/52ff194441ff4353afd2443b1aae1024 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcee58b23e99428eb465117124372446 : SUCCESS in 2m 52s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/d1b000ab07144bc8809750488cd13305 : SUCCESS in 28m 23s","accounts_in_message":[],"_revision_number":3},{"id":"b5a9de4332ec816c78c263da54afb8502e1f90e6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 11:29:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8f7b654fe012b4715b400fc39d3ecaba92508019","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 11:33:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea88ec999dc549d0bdd920e308d17c02\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ed4218b47e847cc8483b8c90cf8d864 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3},{"id":"1fd201658ba0f926822afc11bca5e1e4c165fd1b","tag":"autogenerated:gerrit:revert","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-03 20:02:15.000000000","message":"Created a revert of this change as I79df2b9500be5269ecd9894ed4cf46a5edb1592a","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"886e7170f1d543da37de738660efee0a01cf1bac","revisions":{"1141261a283756f87fb664b193b3aa24d37b8aa1":{"kind":"REWORK","_number":1,"created":"2023-04-03 09:06:29.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/83/879283/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/83/879283/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/83/879283/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/83/879283/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/83/879283/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/83/879283/1"}}},"commit":{"parents":[{"commit":"de69de18abab88a5f01655b13d8aed431409afb9","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/de69de18abab88a5f01655b13d8aed431409afb9"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-04-03 09:06:18.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-04-03 09:06:18.000000000","tz":120},"subject":"Moves supported python runtimes from version 3.8 to 3.10","message":"Moves supported python runtimes from version 3.8 to 3.10\n\nWithin 2023.2 python version 3.9 and 3.10 are the\nsupported python runtimes [1].\n[1] https: //review.opendev.org/c/openstack/governance/+/872232\n\nChange-Id: I47b536149e8119324b3b715b2103b803b74a8a56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/1141261a283756f87fb664b193b3aa24d37b8aa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/1141261a283756f87fb664b193b3aa24d37b8aa1"}]},"branch":"refs/heads/master"},"1ea9c7dfe93df204eb006a2319f24d2df8eb44eb":{"kind":"REWORK","_number":2,"created":"2023-04-10 01:54:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/879283/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/83/879283/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/83/879283/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/83/879283/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/83/879283/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/83/879283/2"}}},"commit":{"parents":[{"commit":"de69de18abab88a5f01655b13d8aed431409afb9","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/de69de18abab88a5f01655b13d8aed431409afb9"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-04-03 09:06:18.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-10 01:54:02.000000000","tz":0},"subject":"Moves supported python runtimes from version 3.8 to 3.10","message":"Moves supported python runtimes from version 3.8 to 3.10\n\nWithin 2023.2 python version 3.9 and 3.10 are the\nsupported python runtimes [1].\n[1] https: //review.opendev.org/c/openstack/governance/+/872232\n\nChange-Id: I47b536149e8119324b3b715b2103b803b74a8a56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/1ea9c7dfe93df204eb006a2319f24d2df8eb44eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/1ea9c7dfe93df204eb006a2319f24d2df8eb44eb"}]},"branch":"refs/heads/master"},"886e7170f1d543da37de738660efee0a01cf1bac":{"kind":"REWORK","_number":3,"created":"2023-04-10 04:06:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/879283/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/83/879283/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/83/879283/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/83/879283/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/83/879283/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/83/879283/3"}}},"commit":{"parents":[{"commit":"de69de18abab88a5f01655b13d8aed431409afb9","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/de69de18abab88a5f01655b13d8aed431409afb9"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-04-03 09:06:18.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-10 04:06:20.000000000","tz":540},"subject":"Moves supported python runtimes from version 3.8 to 3.10","message":"Moves supported python runtimes from version 3.8 to 3.10\n\nWithin 2023.2 python version 3.9 and 3.10 are the\nsupported python runtimes [1].\n[1] https: //review.opendev.org/c/openstack/governance/+/872232\n\nChange-Id: I47b536149e8119324b3b715b2103b803b74a8a56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/886e7170f1d543da37de738660efee0a01cf1bac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/886e7170f1d543da37de738660efee0a01cf1bac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
