)]}'
{"id":"openstack%2Foslo.limit~906608","triplet_id":"openstack%2Foslo.limit~master~I3c40ded7174691473e4ce36eb37da73707178f3c","project":"openstack/oslo.limit","branch":"master","topic":"bump-hacking","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-26 01:01:33.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-01-27 11:30:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3c40ded7174691473e4ce36eb37da73707178f3c","subject":"Bump hacking","status":"MERGED","created":"2024-01-25 03:13:23.000000000","updated":"2024-01-27 11:31:01.000000000","submitted":"2024-01-27 11:30:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906608-bump-hacking","meta_rev_id":"654a7f8c76b35f88f17c57b90ea40ec315737285","_number":906608,"virtual_id_number":906608,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-27 11:30:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-01-27 09:46:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-01-27 09:46:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-25 05:34:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-25 06:09:08.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"d6345df3895d02df8bde54386da64d80252995d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-25 03:13:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05e642ae4d375bbd673f5b53b1fa44a3ce3716c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 05:34:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f5359fb357343d1a56a7f3c4df29f83\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/928ae52482a743b7894186c9a9c22590 : SUCCESS in 2m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b74e4a43de54127a29c6210266f5b5b : SUCCESS in 3m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/36d199d0b7374bd6abaac4a8fcdc772f : SUCCESS in 1h 26m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/839d4dd10de0469ba35a07f28da23029 : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aca88c60978b47ef958d9dcb81b27cc6 : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7fb41e3439234787b839421b1e34c665 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63d0e1aa73a84d778e787550ab09a6ad : SUCCESS in 3m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c295ba4930d24ee291a3cd091e13cace : SUCCESS in 4m 20s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/13e829a58f9c4c749bc654857dfdcfbd : SUCCESS in 30m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d9ac9de7195b22a1fc4698e7f4dd42d46f6d85ac","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-01-25 06:09:08.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6612bf8f231877ccc65771721c0316076759c7b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-25 16:12:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2f661aa5266f098360f5170951ddfd5db9e3577c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 17:33:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d81b0fd1e1648fc9c63f6038e84066a\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/92995f60833843e1a2e3e7b958e1cbc0 : SUCCESS in 2m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b76ac9f99e0f4890aacf57f83e916941 : SUCCESS in 3m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7f7089dedf104bb0957a3d2eb851cc23 : FAILURE in 1h 10m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad9f5f82e55a403fa15a89e9c6b4e1ea : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46b8ed2bacfd43e3ad3fd608f60d6395 : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b07d53895bfd44c1bec3ae2275605a55 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e54435bd532b4d0ea7d3865ded7f9d97 : SUCCESS in 3m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1e78792826947d2a89ab973c695d395 : SUCCESS in 3m 02s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/19b6055f4f524321871df5c090868f47 : SUCCESS in 31m 57s","accounts_in_message":[],"_revision_number":2},{"id":"d51d09fa4b9bf101b44a6d7402ffbbe7857f0ca3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-26 01:01:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5ea12d7efc4681470fb2fc6d28db764468a979b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 02:36:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abc5d5f190734df78b433c5fd91db43c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/31dcc44dd85a40bfbd24b98b59be4dfd : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6fa15c695034ee7914e420a162f39f4 : SUCCESS in 2m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/084d18308ac34a98afd6e98fa1765e3c : SUCCESS in 1h 28m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48538eb578c542dfa0d47b5e020fb68b : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc281ab63ff544e891a84ce3a7a93309 : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e2aeb00b9f18429285eb4244143e3207 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de927de99c5a48d9a3efd12d2ee3dfcc : SUCCESS in 3m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b75dee99eba04e849d770e586d40675b : SUCCESS in 2m 39s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/8e13a82b448e46908dbc6368f28f3dca : SUCCESS in 32m 33s","accounts_in_message":[],"_revision_number":2},{"id":"8474a6dae852d06c389e8b8d874652d0ff1ad474","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-27 09:46:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ba7ffdb737993c19200e40a3a86a2fb3caf216f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-27 09:46:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"744ea88505cd7ef9c7e61b5879cf2f89c838bb70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-27 11:30:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e2232eee7a0480195a420f958726a93\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c8373a89d4054dbf9791daac9936f509 : SUCCESS in 1m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/75b8a02b4c53485a8f44c13a7e5d3127 : SUCCESS in 1h 37m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fff863a3bcc24615b98648109aabd915 : SUCCESS in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de05293a354a4f59bedaa230e47399f2 : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dcdba55f19be4a8abafb46a7219087a9 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebba4d0bc25045ac8bbe27c86ec9b70a : SUCCESS in 2m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fad531cc9da4eaa85cdd944e2c1e0a3 : SUCCESS in 2m 50s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/ca56e6f2827c4189aaee177048b4bdbb : SUCCESS in 29m 14s","accounts_in_message":[],"_revision_number":2},{"id":"d3cb7152a4957ea1f55f51061e698d12591677b3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-27 11:30:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"654a7f8c76b35f88f17c57b90ea40ec315737285","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-27 11:31:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81b1b822af854705ac33c6d3437ab597\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b8b14bc619741528890eaec8742af52 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/831a364f304d4d07a165b16ff0b0a4c5 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d85a3ad38c3069b5c20a4eeaa29c253fed722ddc","revisions":{"a28571538a9d59d445501d06d0efaa322e884f90":{"kind":"REWORK","_number":1,"created":"2024-01-25 03:13:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/906608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/08/906608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/08/906608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/08/906608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/08/906608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/08/906608/1"}}},"commit":{"parents":[{"commit":"ba8b9aba0b918f0c3c73f0d7a3a1f18f4c341be8","subject":"Ensure endpoint_id is set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/ba8b9aba0b918f0c3c73f0d7a3a1f18f4c341be8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-25 03:11:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-25 03:11:19.000000000","tz":540},"subject":"Bump hacking","message":"Bump hacking\n\nhacking 3.0.x is too old.\n\nChange-Id: I3c40ded7174691473e4ce36eb37da73707178f3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/a28571538a9d59d445501d06d0efaa322e884f90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/a28571538a9d59d445501d06d0efaa322e884f90"}]},"branch":"refs/heads/master"},"d85a3ad38c3069b5c20a4eeaa29c253fed722ddc":{"kind":"REWORK","_number":2,"created":"2024-01-25 16:12:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/906608/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/08/906608/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/08/906608/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/08/906608/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/08/906608/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/08/906608/2"}}},"commit":{"parents":[{"commit":"ba8b9aba0b918f0c3c73f0d7a3a1f18f4c341be8","subject":"Ensure endpoint_id is set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/ba8b9aba0b918f0c3c73f0d7a3a1f18f4c341be8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-25 03:11:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-25 16:12:45.000000000","tz":540},"subject":"Bump hacking","message":"Bump hacking\n\nhacking 3.0.x is too old.\n\nChange-Id: I3c40ded7174691473e4ce36eb37da73707178f3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/d85a3ad38c3069b5c20a4eeaa29c253fed722ddc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/d85a3ad38c3069b5c20a4eeaa29c253fed722ddc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
