)]}'
{"id":"openstack%2Foslo.limit~932787","triplet_id":"openstack%2Foslo.limit~master~I1f005fa2c925612ebfd97d0ca9c2727a4d162523","project":"openstack/oslo.limit","branch":"master","topic":"pyupgrade","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-31 12:48:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1f005fa2c925612ebfd97d0ca9c2727a4d162523","subject":"Run pyupgrade to clean up Python 2 syntaxes","status":"MERGED","created":"2024-10-19 14:15:24.000000000","updated":"2024-10-31 12:49:24.000000000","submitted":"2024-10-31 12:48:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932787-pyupgrade","meta_rev_id":"c522747007f89c07c2f2adb996184f8e284fded6","_number":932787,"virtual_id_number":932787,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-31 12:48:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2024-10-21 13:03:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-31 10:57:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-31 10:57:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-19 15:31:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-21 13:03:05.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-10-31 10:57:11.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"0c985bfce5526b607310005e887457a276a19345","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-19 14:15:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7aa05123aefa7b090f600c9fe6817ff67cb0dc77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-19 15:31:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b925579ffe5a4faeb2b88f608fa978d3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1cf93c4162d4c11a5ea689413f7442c : SUCCESS in 3m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7dc1e7b002754bb1b8edc3d5afb8794e : FAILURE in 1h 07m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cd8aba7584e4f9a9bf4f2c0380db81e : SUCCESS in 2m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad928fe119ab4a59be1e269771a4cfe3 : SUCCESS in 2m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a8523c79375d48a39e5fa90ef49e67bc : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e80abe6f55e34456830061600da3fed2 : SUCCESS in 2m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77e9f1b4134f430cbe71b265e119f468 : SUCCESS in 2m 12s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/d741a845666e4cc482a0ce06b14699ee : SUCCESS in 25m 40s","accounts_in_message":[],"_revision_number":1},{"id":"961eaab07aef6fa6f21c9c3bfe0bfea6dbd82aa2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-21 10:46:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ce93cf8bdfe538dc7e7b6193cfe5e8ff3317484c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-21 10:50:36.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"4e0b03b28885b8afe3868f7cb647ced99f8ac14e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 12:34:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64b31af3cbe84f78bce492cad8739fbc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ff486a17c994ba39d18e12daa2bf295 : SUCCESS in 2m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d282d599ab574d0fbc966bb96cdccb2d : SUCCESS in 1h 38m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/158f7b2782254a648c8b796869dcadef : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5a2f5b22c624a3faa2ab9f0206adea3 : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bc84389ae7c64162be54523ce19b4af7 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdf627dfcdd74242b14b610bcf11650e : SUCCESS in 2m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23026243c50e4b7eac7b30caa120eb3a : SUCCESS in 2m 15s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/23fe2b210cbd4cd8bcd580dd98312840 : SUCCESS in 17m 19s","accounts_in_message":[],"_revision_number":3},{"id":"c0875aafef4ef9c449fa48d8e95c7393581d9baf","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-10-21 13:03:05.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8ef0075b3f7d50e0994f6efd8939777a10a4965b","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-10-31 10:57:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c93247dcb82b44029ed78741233b59c8d2e3d9a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 10:58:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7eca2b532385a73fb443a8adf8897b943e5c8c55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 12:48:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e2ee17b6b17457bae6a58a1078239c1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c538f7fe4d5349dba71a567fe15dec07 : SUCCESS in 1h 41m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f3e2aff7c2243d19af26555b86133c7 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6400be2178504d4d8a9dedd94e2d9775 : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cdaa9bb35bb7456299f2c2a781020123 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d10626c508f42369e586756df6ff9f6 : SUCCESS in 2m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e23b4fa93cda40238fac552efbad6594 : SUCCESS in 2m 47s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/cc933a56c23a47f8a466e09f20190d4b : SUCCESS in 19m 18s","accounts_in_message":[],"_revision_number":3},{"id":"797cc241925c54b0017c1a69bb02477c83441ef9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 12:48:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"c522747007f89c07c2f2adb996184f8e284fded6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 12:49:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e4eeeb629d34336981648bfa6d7f496\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6a0ed7fa75a42c89b99f1522dd6a9ad : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d463a207fa1e4f0da9dbe516e3fcb5bb : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"585d49fa302e7d1132e62ef6cf4c0b1a9299e5f2","revisions":{"89b3d7ad0d32642ec1370a64e6b92738d1dcab7c":{"kind":"REWORK","_number":1,"created":"2024-10-19 14:15:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/932787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/87/932787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/87/932787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/87/932787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/87/932787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/87/932787/1"}}},"commit":{"parents":[{"commit":"bc1b59096694e03b17cd7f6a7df22e0854eb9ba1","subject":"Declare Python 3.12 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/bc1b59096694e03b17cd7f6a7df22e0854eb9ba1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-19 14:15:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-19 14:15:13.000000000","tz":540},"subject":"Run pyupgrade to clean up Python 2 syntaxes","message":"Run pyupgrade to clean up Python 2 syntaxes\n\nUpdate all .py source files by\n $ pyupgrade --py3-only $(git ls-files | grep \".py$\")\nto modernize the code according to Python 3 syntaxes.\n\nChange-Id: I1f005fa2c925612ebfd97d0ca9c2727a4d162523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/89b3d7ad0d32642ec1370a64e6b92738d1dcab7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/89b3d7ad0d32642ec1370a64e6b92738d1dcab7c"}]},"branch":"refs/heads/master"},"df1b6f2043c121eace9c019fe999078ccd413ac4":{"kind":"REWORK","_number":2,"created":"2024-10-21 10:46:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/932787/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/87/932787/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/87/932787/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/87/932787/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/87/932787/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/87/932787/2"}}},"commit":{"parents":[{"commit":"14ea68e4b88a5ee8861a8581c52a9b7216aaaec8","subject":"pre-commit: Bump versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/14ea68e4b88a5ee8861a8581c52a9b7216aaaec8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-19 14:15:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-21 10:46:16.000000000","tz":540},"subject":"Run pyupgrade to clean up Python 2 syntaxes","message":"Run pyupgrade to clean up Python 2 syntaxes\n\nUpdate all .py source files by\n $ pyupgrade --py3-only $(git ls-files | grep \".py$\")\nto modernize the code according to Python 3 syntaxes.\n\nChange-Id: I1f005fa2c925612ebfd97d0ca9c2727a4d162523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/df1b6f2043c121eace9c019fe999078ccd413ac4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/df1b6f2043c121eace9c019fe999078ccd413ac4"}]},"branch":"refs/heads/master"},"585d49fa302e7d1132e62ef6cf4c0b1a9299e5f2":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-10-21 10:50:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/932787/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/87/932787/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/87/932787/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/87/932787/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/87/932787/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/87/932787/3"}}},"commit":{"parents":[{"commit":"14ea68e4b88a5ee8861a8581c52a9b7216aaaec8","subject":"pre-commit: Bump versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/14ea68e4b88a5ee8861a8581c52a9b7216aaaec8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-19 14:15:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-21 10:50:36.000000000","tz":0},"subject":"Run pyupgrade to clean up Python 2 syntaxes","message":"Run pyupgrade to clean up Python 2 syntaxes\n\nUpdate all .py source files by\n $ pyupgrade --py3-only $(git ls-files | grep \".py$\")\nto modernize the code according to Python 3 syntaxes.\n\nAlso add the pyupgrade hook to pre-commit to avoid merging additional\nPython 2 syntaxes.\n\nChange-Id: I1f005fa2c925612ebfd97d0ca9c2727a4d162523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/585d49fa302e7d1132e62ef6cf4c0b1a9299e5f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/585d49fa302e7d1132e62ef6cf4c0b1a9299e5f2"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
