)]}'
{"id":"openstack%2Foslo.limit~933535","triplet_id":"openstack%2Foslo.limit~master~I506c536b601cfab773f6c49cde9a6d91a7854f4b","project":"openstack/oslo.limit","branch":"master","hashtags":[],"change_id":"I506c536b601cfab773f6c49cde9a6d91a7854f4b","subject":"Bump min openstacksdk to use name filter for service","status":"MERGED","created":"2024-10-27 05:04:46.000000000","updated":"2024-10-27 13:21:46.000000000","submitted":"2024-10-27 13:20:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933535","meta_rev_id":"f238169b73d7a41784acf7d00675febe78e3fe8e","_number":933535,"virtual_id_number":933535,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-27 13:20:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-10-27 11:48:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-10-27 11:48:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-27 07:10:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e3cdb991c8355c2433f0d66b7ae8b46d097473b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-27 05:04:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a05175f9350942fc20f21c258311657e6582710d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-27 07:10:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a784eec96fd41c1b64db53a1790d6d5\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7a21d7dca99e4081b091745c6341375a : SUCCESS in 1m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31af987b7c89440a829ccc63072c0d7a : SUCCESS in 2m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3ceadf3628ab44da90baaae1cc1aefb3 : SUCCESS in 1h 39m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb54364cd4f8477cb0e69dc25664fc8e : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ea3042f3175483fbd76efba778e6dcc : SUCCESS in 3m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5d6e52dbe89b4f0c85729d275009cbe4 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e3d9ca690e8477993c39eb61c7ef63e : SUCCESS in 2m 35s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/7694f3390dc54203b950e25075eef60a : SUCCESS in 21m 38s","accounts_in_message":[],"_revision_number":1},{"id":"602bb987481dd456147f8bc5b17ff854dba2499f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-27 11:48:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1d0a08e80d187b9782a9cbe2e7c11ffc725566a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-27 11:49:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d234230ec7c1a9fa8363a81bbcb601f1f4c7329a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-27 13:20:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d3c67a5b8d74b96aa61a7391562da5f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/fec420f5ce334dbb9da1aa69905a1290 : SUCCESS in 1m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/86b7b52a46eb4cd79ee64c32b9594879 : SUCCESS in 1h 28m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fd52ed226bd403f857b7042ae613923 : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfa40dd2338b49809b06eda65439557c : SUCCESS in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ed439c96a7064e2987eb416b730cf97e : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bc451e9ab884a70875422c749056453 : SUCCESS in 3m 47s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/5a748dfeaaf3494d92d898230afcdfcb : SUCCESS in 14m 29s","accounts_in_message":[],"_revision_number":1},{"id":"3da5a0f53d8ba8d2b08002e37bcc26ac8d8f0f1d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-27 13:20:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f238169b73d7a41784acf7d00675febe78e3fe8e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-27 13:21:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25807c14ab9f412a999895e3506d0753\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81841c67a993424aadc05dc45bbb1e51 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fd330a22ac8250f96647d9831dce3e7df9e0b370","revisions":{"fd330a22ac8250f96647d9831dce3e7df9e0b370":{"kind":"REWORK","_number":1,"created":"2024-10-27 05:04:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/933535/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/35/933535/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/35/933535/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/35/933535/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/35/933535/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/35/933535/1"}}},"commit":{"parents":[{"commit":"a9f339bd48da1bf78cabc11a0ed7029465e76449","subject":"Merge \"pre-commit: Bump versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/a9f339bd48da1bf78cabc11a0ed7029465e76449"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-27 05:03:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-27 05:03:12.000000000","tz":540},"subject":"Bump min openstacksdk to use name filter for service","message":"Bump min openstacksdk to use name filter for service\n\nAs follow-up of 9575a24796e019fd66f1bb1a5ef0bcbfc167351a , bump\nthe minimum openstacksdk version because name filter in service list is\nnot available in the current min version[1].\n\n[1] 1b21b7bda1ead2d28568acea7fc0c0addaa913b0\n\nChange-Id: I506c536b601cfab773f6c49cde9a6d91a7854f4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/fd330a22ac8250f96647d9831dce3e7df9e0b370"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/fd330a22ac8250f96647d9831dce3e7df9e0b370"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
