)]}'
{"id":"openstack%2Foslo.limit~947902","triplet_id":"openstack%2Foslo.limit~master~Ic28824a7b9395aab4837bf9b165d21ff0d856029","project":"openstack/oslo.limit","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-05-01 01:55:55.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic28824a7b9395aab4837bf9b165d21ff0d856029","subject":"Fix wrong argument to determine identity endpoint","status":"ABANDONED","created":"2025-04-23 04:53:49.000000000","updated":"2025-05-01 01:55:55.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9b305bf39630fb6f9afb91ef1bb95aaade628c70","_number":947902,"virtual_id_number":947902,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-23 09:58:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-23 05:49:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"478b72d8d612d1fa1333cfd5e1c5d29453351cc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-23 04:53:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4f206419614205149f1608d733d5f9154daa205","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-23 05:00:01.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"db348c965c3b3e093cedc8b6023aa569a07fa3b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 05:49:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ec0020434a84b4fadc99976c87a9228\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1546f68fca1440e6ab1a8f1faa48ed49 : SUCCESS in 2m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5db0af45f2374b71aa22d782ed2af044 : SUCCESS in 54m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/288103d216ad4686821b9a9c17a075ba : SUCCESS in 1m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df0bfd76e3c64c7eb2e57c52648c8daf : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e28924cb9a254a26aa14a09bc5ff219e : SUCCESS in 2m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/56d8f66a08854567adb00f358512959f : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/775b7d1aafb1454091234280d2200dd1 : SUCCESS in 2m 27s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/7959c2e97be74d488cdade2479e09c28 : SUCCESS in 17m 04s","accounts_in_message":[],"_revision_number":1},{"id":"061d352e5ca4dbafa063d9612da8a3ea1163575d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-23 09:05:58.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b566bde4764fff858f9e927bb27cf503621eedde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 09:58:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f7c838f1e634fc8b4af1737b73de2a0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/771f63e2cdcf4b5f94c074c3a0f1fbb4 : SUCCESS in 1m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/07c93a57a9814141a73f6d65dcf6c779 : SUCCESS in 52m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e69a86ac6014e56b20f682f7400dd65 : SUCCESS in 1m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66fb26eaa379438783e5d458a75a5299 : SUCCESS in 1m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/57144f03ce2b4ae3a5b10f685fbaeec8 : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/166fdf457257431db43444075698f775 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/884921fb617844f1985e169ac2267504 : SUCCESS in 3m 12s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/93541080db59413bae21da957a58163e : SUCCESS in 22m 22s","accounts_in_message":[],"_revision_number":2},{"id":"9b305bf39630fb6f9afb91ef1bb95aaade628c70","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-01 01:55:55.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"75d720923dbcb0b9a93840bcaacb533f14ca3449","revisions":{"9cd297e49c4e9a483d734742e7b70ab34610fdd4":{"kind":"REWORK","_number":1,"created":"2025-04-23 04:53:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/947902/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/02/947902/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/02/947902/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/02/947902/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/02/947902/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/02/947902/1"}}},"commit":{"parents":[{"commit":"5be2965266370531925f75521cfd1078063903f0","subject":"Merge \"tox: Remove basepython\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/5be2965266370531925f75521cfd1078063903f0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-23 04:53:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-23 04:53:15.000000000","tz":540},"subject":"Fix wrong argument to determine identity endpoint","message":"Fix wrong argument to determine identity endpoint\n\n... from which the limit information is pulled.\n\nChange-Id: Ic28824a7b9395aab4837bf9b165d21ff0d856029\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/9cd297e49c4e9a483d734742e7b70ab34610fdd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/9cd297e49c4e9a483d734742e7b70ab34610fdd4"}]},"branch":"refs/heads/master"},"75d720923dbcb0b9a93840bcaacb533f14ca3449":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-04-23 09:05:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/947902/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/02/947902/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/02/947902/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/02/947902/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/02/947902/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/02/947902/2"}}},"commit":{"parents":[{"commit":"e33cba9b55febec3fcd269d2b868e0bdf2642d64","subject":"Remove unused adapter options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/e33cba9b55febec3fcd269d2b868e0bdf2642d64"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-23 04:53:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-23 09:05:58.000000000","tz":0},"subject":"Fix wrong argument to determine identity endpoint","message":"Fix wrong argument to determine identity endpoint\n\n... from which the limit information is pulled.\n\nChange-Id: Ic28824a7b9395aab4837bf9b165d21ff0d856029\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/75d720923dbcb0b9a93840bcaacb533f14ca3449"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/75d720923dbcb0b9a93840bcaacb533f14ca3449"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
