)]}'
{"id":"openstack%2Foslo.limit~961063","triplet_id":"openstack%2Foslo.limit~master~I366a755863a09c39e0da6006f7eff30dcead94c9","project":"openstack/oslo.limit","branch":"master","hashtags":[],"change_id":"I366a755863a09c39e0da6006f7eff30dcead94c9","subject":"Add repository URL","status":"MERGED","created":"2025-09-14 08:10:12.000000000","updated":"2025-09-14 15:00:34.000000000","submitted":"2025-09-14 14:59:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961063","meta_rev_id":"a3730a4e5f77e9d23341cb59e02c28da87088d12","_number":961063,"virtual_id_number":961063,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-14 14:59:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-09-14 11:14:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-09-14 11:14:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-14 10:04:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6088aab167b5821609721959824ceeb0a1b70517","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-14 08:10:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c17596885fcc0bf5e95ed7f6db84568c021459f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-14 08:12:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"159e3c22c58fa107c4fb02c1f7f7f239d6a1eff3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-14 10:04:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0dd012d67c446f4a4b4cc467686357f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d6151faf3e74279b532cf10875aea95 : SUCCESS in 2m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/11a90e3a0b85459e9cb0445a190ef194 : SUCCESS in 1h 47m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65bc8db50d4e4306be9543ded55c5e0e : SUCCESS in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a94f0a31c2e74995b9a68e935cc61142 : SUCCESS in 2m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ea9026f24cd941599d1ee4b2dd418f44 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4e63366f0fc74c67902b8daa37487f2b : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03b13f00e46842a799947bfb229e2640 : SUCCESS in 2m 47s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/3d4c6fcb78f84a839422c3503c6a857d : SUCCESS in 25m 40s","accounts_in_message":[],"_revision_number":2},{"id":"dee8e7e17b82273927fe7cf64ffa989a9758f9c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-14 11:14:02.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"1d1a80be35c4391246946c9ccb7cc11919328fda","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-14 11:14:06.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dcd4b8f9994663e4f7798758ed734bd3d7f53b6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-14 13:02:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63d773064e514f94931d0351afe40b50\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27b86ddd9dcb4d1ab6c7cdd882eb976e : SUCCESS in 6m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55fb2b8fe2d3448bbf45894298dc1963 : SUCCESS in 1h 43m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd0177bc315b46f499cd9acf44d2b6e0 : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be0e1476754f414eaed12b0756570ed8 : SUCCESS in 3m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/17a6d3ae87284445850e2a463e200d43 : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/307519563e444ef2a5aa6db0ba58f9af : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9929c621eaa54ccd8826564f3948f123 : SUCCESS in 3m 21s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/9f2bdabb30c842bf87e972e4cfb60e78 : SUCCESS in 15m 28s","accounts_in_message":[],"_revision_number":3},{"id":"e66d0f6ae97372a14f2dbd7a59a538b89542aab8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-14 13:02:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"47b9f0a67ec2f296e63f74e554fff12169fcd9d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-14 14:59:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d03b92eda14647c0b7050e4e100b67f8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e02dc5183b44283b7a3567d80bad38f : SUCCESS in 1h 52m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8b81761af844b3580fa9fec20d376c1 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1731027715484ac6aa8318b0869c5d4b : SUCCESS in 6m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6763d0aa4136472080b5e31cce9effdb : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c06bf3f6200419193efe0d14ed0f9ea : SUCCESS in 3m 18s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/f5910bef91d94dc497ff8b377aacaa64 : SUCCESS in 25m 16s","accounts_in_message":[],"_revision_number":3},{"id":"e000e30de8a0758c0a3e146a144df0fb1a74afb8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-14 14:59:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a3730a4e5f77e9d23341cb59e02c28da87088d12","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-14 15:00:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61c24ac1bc034a5d89b9774f57d48975\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50c5f91358334c768f4b732d1bbdee52 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3bb99cf441110abaada73fddd3cbd599be856b64","revisions":{"c3481bee9038f022d88befaeb8dc70e9bdac2c0c":{"kind":"REWORK","_number":1,"created":"2025-09-14 08:10:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/961063/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/63/961063/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/63/961063/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/63/961063/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/63/961063/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/63/961063/1"}}},"commit":{"parents":[{"commit":"b1098227a10019941eadf19eb2d2ba2c27f200b3","subject":"pre-commit: Bump dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/b1098227a10019941eadf19eb2d2ba2c27f200b3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-14 08:09:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-14 08:10:04.000000000","tz":540},"subject":"Add repository URL","message":"Add repository URL\n\nChange-Id: I366a755863a09c39e0da6006f7eff30dcead94c9\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/c3481bee9038f022d88befaeb8dc70e9bdac2c0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/c3481bee9038f022d88befaeb8dc70e9bdac2c0c"}]},"branch":"refs/heads/master"},"bb7bd8cab4db988e1dc9b21a0ede8023bd9c5036":{"kind":"REWORK","_number":2,"created":"2025-09-14 08:12:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/961063/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/63/961063/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/63/961063/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/63/961063/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/63/961063/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/63/961063/2"}}},"commit":{"parents":[{"commit":"b1098227a10019941eadf19eb2d2ba2c27f200b3","subject":"pre-commit: Bump dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/b1098227a10019941eadf19eb2d2ba2c27f200b3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-14 08:09:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-14 08:11:57.000000000","tz":540},"subject":"Add repository URL","message":"Add repository URL\n\nChange-Id: I366a755863a09c39e0da6006f7eff30dcead94c9\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/bb7bd8cab4db988e1dc9b21a0ede8023bd9c5036"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/bb7bd8cab4db988e1dc9b21a0ede8023bd9c5036"}]},"branch":"refs/heads/master"},"3bb99cf441110abaada73fddd3cbd599be856b64":{"kind":"REWORK","_number":3,"created":"2025-09-14 11:14:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/961063/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/63/961063/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/63/961063/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/63/961063/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/63/961063/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/63/961063/3"}}},"commit":{"parents":[{"commit":"b1098227a10019941eadf19eb2d2ba2c27f200b3","subject":"pre-commit: Bump dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/b1098227a10019941eadf19eb2d2ba2c27f200b3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-14 08:09:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-14 11:14:02.000000000","tz":0},"subject":"Add repository URL","message":"Add repository URL\n\nChange-Id: I366a755863a09c39e0da6006f7eff30dcead94c9\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3bb99cf441110abaada73fddd3cbd599be856b64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3bb99cf441110abaada73fddd3cbd599be856b64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
