)]}'
{"id":"openstack%2Foslo.limit~961329","triplet_id":"openstack%2Foslo.limit~master~Icbf1eb9bab08690a35e402e7a5bd1e0b8912e88d","project":"openstack/oslo.limit","branch":"master","topic":"bug/2123895","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-16 13:40:23.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2025-09-16 15:27:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icbf1eb9bab08690a35e402e7a5bd1e0b8912e88d","subject":"Fix endpoint query","status":"MERGED","created":"2025-09-16 02:28:56.000000000","updated":"2025-09-16 15:27:55.000000000","submitted":"2025-09-16 15:27:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961329-bug/2123895","meta_rev_id":"c6b668f2d4db1f7c49e06bb0eaeb1cf3eaa3c719","_number":961329,"virtual_id_number":961329,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-16 15:27:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-09-16 09:04:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2025-09-16 13:40:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2025-09-16 13:40:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-16 04:31:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-16 07:18:34.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2025-09-16 09:04:17.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"9686f475614dfb208870fcbb84f6415750660e50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-16 02:28:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0028d637d39573b5eeffa63368ca59619bf6479e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 04:31:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7c5cbbd3b294463985ce7c9fb36aa01\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24e1b5a4c453421b9ab99fccf7f3182f : SUCCESS in 3m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/955f545004fa4c0297c902ce31b39be1 : SUCCESS in 1h 24m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4e7c137b693417fb4f960f4eb7388de : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/870621604c68456f90d90c8f38c73346 : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f803055b05d7427880d9c31f5648470e : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1bcd5c46427c441dae4560d256714b4b : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44d3cc1beec14e4594ca4197c12f4768 : SUCCESS in 3m 14s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/c2a73affb73d4c88a34e585574f1615c : SUCCESS in 18m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a32812490b0c07cc71d9b4168048b7518f04e651","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-16 04:48:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"92c93651f5e149573b190052a68e195494a2de5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-16 06:51:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"062469b21f18d905bd400e29e34e24aeb4feda4e","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2025-09-16 07:18:34.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a993b59023df0cc0f962c80286730a40ff480b93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-16 07:28:18.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"4ece7fa9b7bf896d214ff5f223c44bc9cc22fa92","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-16 09:04:17.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9400c59490e97db5ab29fa24233638d466c63dc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 09:10:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb7d1960d9ae4aad92b630caf2b9021b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a53728ed14b485589e8eeaa4425b7c8 : SUCCESS in 5m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2f2e54915d424f7590701752e4f8387d : SUCCESS in 1h 36m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cf3f0da614e41a69c52ba8d9b1fd961 : SUCCESS in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c657794ded59489ab480cefe49694e94 : SUCCESS in 5m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d0218a2d86ac40249c6f7552c532de4f : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c7b343e6620141eebb1b030337c76174 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bb3100507874994aa9492b561d95f8b : SUCCESS in 4m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2095a00ce4734eeda3bc10b8f8a4b7ce : SUCCESS in 3m 54s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/3fa78facea5b4d81876ae4226a156dd9 : SUCCESS in 23m 41s","accounts_in_message":[],"_revision_number":4},{"id":"8aac01a79f1dc4dfd547016e49cf52c0aa1903d2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-16 13:40:23.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e6f75d7dd0bd653f203045b868c9cbbf3ab63b70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 13:40:43.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"bdcd450ac3f0568d0432cc12b20b330c2ce59d88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 15:27:12.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20536249d92949fabd6ffce791e02288\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fcd719489e5a43a8b5f77f097c80a78b : SUCCESS in 1h 44m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52158510fc914efcbfc2fa3a309788b9 : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9cdcd79357cc484c99533da6917cb9f8 : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a1cd573a276c4924a7957767c5af41e0 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e7f32fea992422f8dd53a36055e97eb : SUCCESS in 3m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0558e8e6cb44533b9f2005dd74be338 : SUCCESS in 3m 35s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/917edff21a354420b83127a403b7b089 : SUCCESS in 15m 28s","accounts_in_message":[],"_revision_number":4},{"id":"b51e15df55b8f84ea7a0067dc9d67486005f9a7b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 15:27:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"c6b668f2d4db1f7c49e06bb0eaeb1cf3eaa3c719","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 15:27:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dca708ee9774cb7b0d672278ffffaad\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/699d0f2a4c594824ba603bde3864d296 : SUCCESS in 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/885065a18b1b4be6bc5906687e65bfc2 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8b3edee1902f5b410da1f5ed872c91afb3a7fc60","revisions":{"e5b3c86d4338d0d4d6d3fd0d767d380a5738495a":{"kind":"REWORK","_number":1,"created":"2025-09-16 02:28:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/961329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/29/961329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/29/961329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/29/961329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/29/961329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/29/961329/1"}}},"commit":{"parents":[{"commit":"b1098227a10019941eadf19eb2d2ba2c27f200b3","subject":"pre-commit: Bump dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/b1098227a10019941eadf19eb2d2ba2c27f200b3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-16 02:27:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-16 02:28:49.000000000","tz":540},"subject":"Fix endpoint qurty","message":"Fix endpoint qurty\n\noepnstacksdk returns generators, not lists, so we have to convert\nthe returned values before doing len or index access.\n\nCloses-Bug: #2123895\nChange-Id: Icbf1eb9bab08690a35e402e7a5bd1e0b8912e88d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/e5b3c86d4338d0d4d6d3fd0d767d380a5738495a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/e5b3c86d4338d0d4d6d3fd0d767d380a5738495a"}]},"branch":"refs/heads/master"},"d49d7eab07734dd477a7da4bc94e96ef219da337":{"kind":"REWORK","_number":2,"created":"2025-09-16 04:48:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/961329/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/29/961329/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/29/961329/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/29/961329/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/29/961329/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/29/961329/2"}}},"commit":{"parents":[{"commit":"b1098227a10019941eadf19eb2d2ba2c27f200b3","subject":"pre-commit: Bump dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/b1098227a10019941eadf19eb2d2ba2c27f200b3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-16 02:27:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-16 04:47:38.000000000","tz":540},"subject":"Fix endpoint qurty","message":"Fix endpoint qurty\n\noepnstacksdk returns generators, not lists, so we have to convert\nthe returned values before doing len or index access.\n\nAlso keystone does not accept endpoint_type with URL suffix.\n\nCloses-Bug: #2123895\nChange-Id: Icbf1eb9bab08690a35e402e7a5bd1e0b8912e88d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/d49d7eab07734dd477a7da4bc94e96ef219da337"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/d49d7eab07734dd477a7da4bc94e96ef219da337"}]},"branch":"refs/heads/master"},"be7d9668940982ed9a20fa3933ba7c402f34ed96":{"kind":"REWORK","_number":3,"created":"2025-09-16 06:51:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/961329/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/29/961329/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/29/961329/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/29/961329/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/29/961329/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/29/961329/3"}}},"commit":{"parents":[{"commit":"b1098227a10019941eadf19eb2d2ba2c27f200b3","subject":"pre-commit: Bump dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/b1098227a10019941eadf19eb2d2ba2c27f200b3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-16 02:27:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-16 06:51:04.000000000","tz":540},"subject":"Fix endpoint qurty","message":"Fix endpoint qurty\n\noepnstacksdk returns generators, not lists, so we have to convert\nthe returned values before doing len or index access.\n\nAlso keystone does not accept endpoint_type with URL suffix.\n\nCloses-Bug: #2123895\nChange-Id: Icbf1eb9bab08690a35e402e7a5bd1e0b8912e88d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/be7d9668940982ed9a20fa3933ba7c402f34ed96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/be7d9668940982ed9a20fa3933ba7c402f34ed96"}]},"branch":"refs/heads/master"},"8b3edee1902f5b410da1f5ed872c91afb3a7fc60":{"kind":"REWORK","_number":4,"created":"2025-09-16 07:28:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/961329/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/29/961329/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/29/961329/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/29/961329/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/29/961329/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/29/961329/4"}}},"commit":{"parents":[{"commit":"b1098227a10019941eadf19eb2d2ba2c27f200b3","subject":"pre-commit: Bump dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/b1098227a10019941eadf19eb2d2ba2c27f200b3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-16 02:27:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-16 07:27:53.000000000","tz":540},"subject":"Fix endpoint query","message":"Fix endpoint query\n\noepnstacksdk returns generators, not lists, so we have to convert\nthe returned values before length check or index access.\n\nAlso keystone does not accept endpoint_type with URL suffix.\n\nCloses-Bug: #2123895\nChange-Id: Icbf1eb9bab08690a35e402e7a5bd1e0b8912e88d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/8b3edee1902f5b410da1f5ed872c91afb3a7fc60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/8b3edee1902f5b410da1f5ed872c91afb3a7fc60"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
