)]}'
{"id":"openstack%2Foslo.limit~961332","triplet_id":"openstack%2Foslo.limit~master~I062c785f468f3ed8b6f250b0a4ba77b83706e718","project":"openstack/oslo.limit","branch":"master","topic":"bug/2123895","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-16 14:34:04.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I062c785f468f3ed8b6f250b0a4ba77b83706e718","subject":"DNM: testing ...","status":"ABANDONED","created":"2025-09-16 02:33:59.000000000","updated":"2025-09-16 14:34:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e01e995ac4dc09df6b59650579d3310a5fd78acf","_number":961332,"virtual_id_number":961332,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-09-16 08:57:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-16 04:38:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"82f875a83797c5209bc976c437db1686b53f2762","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-16 02:33:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77ef6750d99a59f5810bc0aaa8b1937dd3894b55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 04:38:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/480976c0106c4e9a9e184d4df2615169\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5777cf9ace54e53a84d298f18c3dcc4 : SUCCESS in 4m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cda09ab18e324b939e942ba83afc6659 : FAILURE in 18m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cce9342dd1c443cea840a98b36fb155a : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1f73c15ff2c475cb6b6832666defb93 : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2d7017eb9fad40fa9399374a5669d9fd : SUCCESS in 3m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6739d1a622eb4222ad9a82fc476d4780 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5e16e49ee8b46b89f10039f67d9db0c : SUCCESS in 3m 08s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/9490765183aa496388dfb7735830d14a : FAILURE in 24m 45s","accounts_in_message":[],"_revision_number":1},{"id":"c6036187a2bea6d692fae3ed58357cc06bf1a7ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-16 04:48:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b097bf62f4118e5ce4b528dadc9934205219e4c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 06:22:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bb6d3cdcde646e997e7271153077de1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6781262d643046f480b4ec18b588a8ad : SUCCESS in 1h 32m 30s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/bd8f01101f1b413c8be308ccff63b0de : SUCCESS in 23m 29s","accounts_in_message":[],"_revision_number":2},{"id":"290d03ec445b49204f95a869eea567e58373b7ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-16 06:51:16.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c97ed9f2e3f8330b6b80f13399cf589d92262f2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-16 07:28:18.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"1408e040f727a1466aceb99051c2b72043a13791","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 08:57:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c75661eab7434580b2ffd0cf14cbce87\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/086fbe9ff85348a5a5a10091e20f0ee9 : SUCCESS in 1h 19m 05s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/4c23c0de90a0480e8bc762ea7fbaaa02 : SUCCESS in 23m 32s","accounts_in_message":[],"_revision_number":4},{"id":"e01e995ac4dc09df6b59650579d3310a5fd78acf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-16 14:34:04.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"531c58c8557a806999fd4258589d3b3867cf64e7","revisions":{"4666a14e89b317e24acbe2da0f1a37031424ff79":{"kind":"REWORK","_number":1,"created":"2025-09-16 02:33:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/961332/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/32/961332/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/32/961332/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/32/961332/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/32/961332/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/32/961332/1"}}},"commit":{"parents":[{"commit":"e5b3c86d4338d0d4d6d3fd0d767d380a5738495a","subject":"Fix endpoint qurty","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/e5b3c86d4338d0d4d6d3fd0d767d380a5738495a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-16 02:33:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-16 02:33:48.000000000","tz":540},"subject":"DNM: testing ...","message":"DNM: testing ...\n\nDepends-on: https://review.opendev.org/c/openstack/devstack/+/961328\nChange-Id: I062c785f468f3ed8b6f250b0a4ba77b83706e718\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/4666a14e89b317e24acbe2da0f1a37031424ff79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/4666a14e89b317e24acbe2da0f1a37031424ff79"}]},"branch":"refs/heads/master"},"ac80a7400ed36a79bfb3483a264e113a67af3ee3":{"kind":"REWORK","_number":2,"created":"2025-09-16 04:48:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/961332/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/32/961332/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/32/961332/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/32/961332/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/32/961332/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/32/961332/2"}}},"commit":{"parents":[{"commit":"d49d7eab07734dd477a7da4bc94e96ef219da337","subject":"Fix endpoint qurty","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/d49d7eab07734dd477a7da4bc94e96ef219da337"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-16 02:33:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-16 04:48:17.000000000","tz":540},"subject":"DNM: testing ...","message":"DNM: testing ...\n\nDepends-on: https://review.opendev.org/c/openstack/devstack/+/961328\nChange-Id: I062c785f468f3ed8b6f250b0a4ba77b83706e718\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/ac80a7400ed36a79bfb3483a264e113a67af3ee3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/ac80a7400ed36a79bfb3483a264e113a67af3ee3"}]},"branch":"refs/heads/master"},"8bea2c6589500670f485942199bd27808dbdeb1d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-09-16 06:51:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/961332/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/32/961332/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/32/961332/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/32/961332/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/32/961332/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/32/961332/3"}}},"commit":{"parents":[{"commit":"be7d9668940982ed9a20fa3933ba7c402f34ed96","subject":"Fix endpoint qurty","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/be7d9668940982ed9a20fa3933ba7c402f34ed96"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-16 02:33:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-16 06:51:09.000000000","tz":540},"subject":"DNM: testing ...","message":"DNM: testing ...\n\nDepends-on: https://review.opendev.org/c/openstack/devstack/+/961328\nChange-Id: I062c785f468f3ed8b6f250b0a4ba77b83706e718\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/8bea2c6589500670f485942199bd27808dbdeb1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/8bea2c6589500670f485942199bd27808dbdeb1d"}]},"branch":"refs/heads/master"},"531c58c8557a806999fd4258589d3b3867cf64e7":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-09-16 07:28:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/961332/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/32/961332/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/32/961332/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/32/961332/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/32/961332/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/32/961332/4"}}},"commit":{"parents":[{"commit":"8b3edee1902f5b410da1f5ed872c91afb3a7fc60","subject":"Fix endpoint query","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/8b3edee1902f5b410da1f5ed872c91afb3a7fc60"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-16 02:33:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-16 07:28:09.000000000","tz":540},"subject":"DNM: testing ...","message":"DNM: testing ...\n\nDepends-on: https://review.opendev.org/c/openstack/devstack/+/961328\nChange-Id: I062c785f468f3ed8b6f250b0a4ba77b83706e718\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/531c58c8557a806999fd4258589d3b3867cf64e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/531c58c8557a806999fd4258589d3b3867cf64e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
