)]}'
{"id":"openstack%2Foslo.limit~961387","triplet_id":"openstack%2Foslo.limit~stable%2F2025.1~Icbf1eb9bab08690a35e402e7a5bd1e0b8912e88d","project":"openstack/oslo.limit","branch":"stable/2025.1","topic":"bug/2123895","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-10-08 17:43:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icbf1eb9bab08690a35e402e7a5bd1e0b8912e88d","subject":"Fix endpoint query","status":"MERGED","created":"2025-09-16 17:03:55.000000000","updated":"2025-10-08 17:44:17.000000000","submitted":"2025-10-08 17:43:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961387-bug/2123895","cherry_pick_of_change":961384,"cherry_pick_of_patch_set":1,"meta_rev_id":"b1974afdc3116393a8aca9d015ca7f3847db4b46","_number":961387,"virtual_id_number":961387,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-08 17:43:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-10-08 15:57:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-10-08 15:57:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-16 18:45:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-08 15:57:58.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"74565ca1df9bdcce978f8ed3fe8bfe9b2972e248","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-16 17:03:55.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.2.","accounts_in_message":[],"_revision_number":1},{"id":"3529b6ecfd46d485cc764256d4e29189f4db1915","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-16 17:04:21.000000000","message":"Topic bug/2123895-stable/2025.1 removed","accounts_in_message":[],"_revision_number":1},{"id":"4767c9262a52e0aa53ad8b1872c7d673f91f449e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-16 17:04:25.000000000","message":"Topic set to bug/2123895","accounts_in_message":[],"_revision_number":1},{"id":"a1cdcc5766a7c10f9ee5b7ca7166b6639dbf1f37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 18:45:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8a70f66d2bf4cecb4ef33339249fd4a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c20203b0f93844d6865f09026877d3d9 : SUCCESS in 4m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d768e63b6ed84213b80dbf4c3b32a041 : SUCCESS in 1h 27m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e62ce76c12a4184b8e4275afb995219 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d50e4e0f5a847608d3b8c1d6708db15 : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/124d090f18124182a1957d764c532446 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e69978b3245405da1f1ddb490e8f277 : SUCCESS in 4m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d396ca015cd54615acfc40b23abaed1f : SUCCESS in 2m 46s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/5905f47c9dc64b0bb016eac82de57f7c : SUCCESS in 19m 03s","accounts_in_message":[],"_revision_number":1},{"id":"0e05072e78356aa6731a1bd4e4dd71769dcbd980","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-08 15:57:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7df66ffd945ab74bf887f98ae6e4df71c23bc478","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 15:58:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ec4112ed300582bf4766ed8ff2fe53834eb8459","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 17:43:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f67f4bf64ad04565989faef88c6db34c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/36d482d1a96a4222a7ddb69ec84cd84a : SUCCESS in 1h 00m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/249f5979229e4b558adb32a57dfd68a7 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1218c186f333450c9524297fc7e153fa : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/18d7d10479854d8f95fe1a8faf38e6ff : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cffe613b7ee4a3a970c845be75f31ad : SUCCESS in 2m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7b046be82f040069cc8cf8ee6dd5459 : SUCCESS in 3m 49s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/4a2232b1b1354740af04c572d8b06cdd : SUCCESS in 25m 40s","accounts_in_message":[],"_revision_number":1},{"id":"3b6b34510e3c9db2b44cd37b03cd41ab47f400a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 17:43:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b1974afdc3116393a8aca9d015ca7f3847db4b46","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 17:44:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70beebc053e94b94a86ef216a1647d64\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f72a418ee73491183f34e773d26b5b2 : SUCCESS in 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cd789b8354446f9b7ef9a4b01f3904f : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"30a358493c5bf4c523c9e6923cfc1f0fa2743b26","revisions":{"30a358493c5bf4c523c9e6923cfc1f0fa2743b26":{"kind":"REWORK","_number":1,"created":"2025-09-16 17:03:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/961387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/87/961387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/87/961387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/87/961387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/87/961387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/87/961387/1"}}},"commit":{"parents":[{"commit":"327977e0991fb292f4de1939f821958d1e8ff0c9","subject":"Fix ignored KSA adapter options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/327977e0991fb292f4de1939f821958d1e8ff0c9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-16 02:27:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-16 17:03:55.000000000","tz":0},"subject":"Fix endpoint query","message":"Fix endpoint query\n\noepnstacksdk returns generators, not lists, so we have to convert\nthe returned values before length check or index access.\n\nAlso keystone does not accept endpoint_type with URL suffix.\n\nCloses-Bug: #2123895\nChange-Id: Icbf1eb9bab08690a35e402e7a5bd1e0b8912e88d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n(cherry picked from commit 8b3edee1902f5b410da1f5ed872c91afb3a7fc60)\n(cherry picked from commit b14c31dbaac99edda8c60a4f5fca49e0cce5acbb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/30a358493c5bf4c523c9e6923cfc1f0fa2743b26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/30a358493c5bf4c523c9e6923cfc1f0fa2743b26"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
