)]}'
{"id":"openstack%2Foslo.limit~961478","triplet_id":"openstack%2Foslo.limit~master~Ic5c6c64ff838cc6cf13cf122921e45451672ab81","project":"openstack/oslo.limit","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-11-18 13:50:43.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-11-18 11:36:34.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2025-11-18 13:50:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic5c6c64ff838cc6cf13cf122921e45451672ab81","subject":"Accept previously accepted format for endpoint_service_type","status":"MERGED","created":"2025-09-17 12:45:39.000000000","updated":"2025-11-18 13:51:31.000000000","submitted":"2025-11-18 13:50:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"961478","meta_rev_id":"211426eb72f89ec8976e66d8ec839b0e7f9046e6","_number":961478,"virtual_id_number":961478,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-18 13:50:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-11-18 11:36:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-11-18 11:36:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-17 13:08:17.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2025-09-17 15:35:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-17 16:55:55.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"}],"messages":[{"id":"fc2aa1dec3bdf9ad4bfcb87493a0ebfc563defcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-17 12:45:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea03b273ad1dcc9ae63cc290aafe6b03b5279db2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-09-17 13:08:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"024bd51756867cf829d1401d8b44c1cd7ef0b91e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-17 13:31:31.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"3d543b0ba491240ec116d7af72f0062a1bedc743","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 15:35:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/198f82aaf9f1470ea4abd55013c3746d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b83da15a8644c2680f2c00aa1d4066c : SUCCESS in 3m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9eaa2d132e464f0f9e15074f386880c4 : SUCCESS in 2h 00m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54ea63eb5e1a4da3a139783af50e48c8 : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3fdd4266c71040f193559cfc3e9c1844 : SUCCESS in 4m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bdeb4bf7aef7484c841e4e69d29ad10b : SUCCESS in 4m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/70d60a07d83544239ce461d904d3de63 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e6ab907ea3e4cf58bd7965a441f994f : SUCCESS in 3m 10s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/bc4607aba3ad42abbacd6266ffe67bef : SUCCESS in 23m 49s","accounts_in_message":[],"_revision_number":2},{"id":"6c3c65bdd9faefd154dc3f0601d5876c6976370a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-17 16:55:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8e38bc6b8c791c4f4a6828aa1822288a77b6fc6a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-17 17:03:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3ed581a0d5466c32c2ff54c18d2c027b5dac1c59","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-17 14:17:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a19cf00559b0df22190680f84d5b00ba8295c4f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 14:58:32.000000000","message":"Patch Set 2: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/oslo.limit for 961478,2","accounts_in_message":[],"_revision_number":2},{"id":"33d56cfec36980e38dc98b4905890033cef07156","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-17 15:17:48.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"bba1805ef5e13e91388f3164b0288d472de77009","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 17:12:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a90e5fd923664431a0482b5063ba6635\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/755f61482faf4c93b6a1d9c75ed6032a : SUCCESS in 3m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5032ac9bab8040cfa49c7d468f5e62d7 : SUCCESS in 1h 49m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bcdb6439e274fb187f8e350a46d5eeb : SUCCESS in 2m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19907149d5f149228800fdbe882b770f : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c680d953e2654acca86e86e56caa185a : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64d131ffc936431aa9142c9b4487af92 : SUCCESS in 2m 11s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/61f563407feb472bb8f1e7bb3377ee8c : SUCCESS in 25m 02s","accounts_in_message":[],"_revision_number":3},{"id":"4cf3ccfa7bdca5998c740cef069165d51bd75b05","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-18 11:36:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"17399d1ff4f1dd35e905c2c77d29720cfcd0bafe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 11:36:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0b526ea897fcd88bab8f0bb38175780660fa3beb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 13:50:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e26571ebd34b4121b93d78278e5965d1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e04721e68ed54e188bf6d6423f09ee96 : SUCCESS in 1h 40m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e557ecd25a943fc886d5869197d545c : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb48e5a2fadb4cef8281c752a3354ed4 : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ec98f9777cba4cf09c80f714d9925fa7 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4b94465f8fe40c589fc13829b708fe0 : SUCCESS in 3m 31s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/7ff0ba2582784cda8fc217a3d301498c : SUCCESS in 12m 06s","accounts_in_message":[],"_revision_number":3},{"id":"1c75058ac704c94c80a2b801ea0adddffce59093","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 13:50:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"211426eb72f89ec8976e66d8ec839b0e7f9046e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 13:51:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3d0627f0fb5483b8eed0749be5a440b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b20c9efc43b4119ae281bf08a63812a : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e852fe2634a53538dd42b48223f10b4af11675b4","revisions":{"b1b5d93478c8a34d7568a9731f1e6e41096d0fe4":{"kind":"REWORK","_number":1,"created":"2025-09-17 12:45:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/961478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/78/961478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/78/961478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/78/961478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/78/961478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/78/961478/1"}}},"commit":{"parents":[{"commit":"914e812bde78e7c88536aeb837c8badf99244dd0","subject":"Merge \"Fix endpoint query\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/914e812bde78e7c88536aeb837c8badf99244dd0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-17 12:44:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-17 12:45:30.000000000","tz":540},"subject":"Accept previously accepted format for endpoint_service_type","message":"Accept previously accepted format for endpoint_service_type\n\nThis is follow-up of 8b3edee1902f5b410da1f5ed872c91afb3a7fc60 and\nensures the format like publicURL, which was previously accepted, is\nstill accepted.\n\nChange-Id: Ic5c6c64ff838cc6cf13cf122921e45451672ab81\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/b1b5d93478c8a34d7568a9731f1e6e41096d0fe4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/b1b5d93478c8a34d7568a9731f1e6e41096d0fe4"}]},"branch":"refs/heads/master"},"b8452d7bff17318fb841952ca13ebe48a932d0c1":{"kind":"REWORK","_number":2,"created":"2025-09-17 13:31:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/961478/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/78/961478/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/78/961478/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/78/961478/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/78/961478/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/78/961478/2"}}},"commit":{"parents":[{"commit":"914e812bde78e7c88536aeb837c8badf99244dd0","subject":"Merge \"Fix endpoint query\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/914e812bde78e7c88536aeb837c8badf99244dd0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-17 12:44:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-17 13:31:23.000000000","tz":540},"subject":"Accept previously accepted format for endpoint_service_type","message":"Accept previously accepted format for endpoint_service_type\n\nThis is follow-up of 8b3edee1902f5b410da1f5ed872c91afb3a7fc60 and\nensures the format like publicURL, which was previously accepted, is\nstill accepted.\n\nChange-Id: Ic5c6c64ff838cc6cf13cf122921e45451672ab81\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/b8452d7bff17318fb841952ca13ebe48a932d0c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/b8452d7bff17318fb841952ca13ebe48a932d0c1"}]},"branch":"refs/heads/master"},"e852fe2634a53538dd42b48223f10b4af11675b4":{"kind":"REWORK","_number":3,"created":"2025-11-17 15:17:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/961478/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/78/961478/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/78/961478/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/78/961478/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/78/961478/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/78/961478/3"}}},"commit":{"parents":[{"commit":"45104432cb15a79c10532e98240dcc8e2c4b6f3d","subject":"Apply ruff","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/45104432cb15a79c10532e98240dcc8e2c4b6f3d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-17 12:44:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-17 15:17:40.000000000","tz":540},"subject":"Accept previously accepted format for endpoint_service_type","message":"Accept previously accepted format for endpoint_service_type\n\nThis is follow-up of 8b3edee1902f5b410da1f5ed872c91afb3a7fc60 and\nensures the format like publicURL, which was previously accepted, is\nstill accepted.\n\nChange-Id: Ic5c6c64ff838cc6cf13cf122921e45451672ab81\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/e852fe2634a53538dd42b48223f10b4af11675b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/e852fe2634a53538dd42b48223f10b4af11675b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
