)]}'
{"id":"openstack%2Foslo.limit~967071","triplet_id":"openstack%2Foslo.limit~master~Icb464584eb0abd9014450a63e4b732eda728c3d7","project":"openstack/oslo.limit","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-11-17 14:00:21.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-11-17 14:58:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icb464584eb0abd9014450a63e4b732eda728c3d7","subject":"Apply ruff","status":"MERGED","created":"2025-11-13 17:28:20.000000000","updated":"2025-11-17 14:59:34.000000000","submitted":"2025-11-17 14:58:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967071-typing","meta_rev_id":"ec7fbf762079616e35be2b40c6e48168cc5a6ae5","_number":967071,"virtual_id_number":967071,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-17 14:58:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2025-11-17 08:56:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-11-17 14:00:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-13 19:26:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-17 08:56:55.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"65c9a6f80a2ddbc84f26a30a03b9cabdbdd868a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-13 17:28:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66c9e1b631ccca6fa7556921a0b1ed6d02008c41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 19:26:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2bddd57ed2547ef96ec1ebfb0a8d0eb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/12433faf5a594f578c23c4381129f50d : SUCCESS in 1m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/471f0384135645caa8d5565f90032556 : SUCCESS in 2m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/472ae5852f344f45afa13d8b71d97443 : SUCCESS in 3m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d2d912cf614e4aed8000230154437179 : SUCCESS in 1h 52m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3dd1e3aecd2471b8ccb2ee3d7faea42 : SUCCESS in 2m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b431241336e435cacbd45b2004c7d92 : SUCCESS in 1m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ded7cc441b32452a8cb449d7de01e8a2 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa543757ca6b45b1b0fe79c4acfbcd85 : SUCCESS in 3m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4e3670b92824c64951d3490565e0b0f : SUCCESS in 2m 28s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/14236e97467c44488e21bf7afd548faa : SUCCESS in 10m 45s","accounts_in_message":[],"_revision_number":1},{"id":"816edf3c4bf2c0bad60bae5adcf02d38b23bfa63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-16 13:11:16.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1cd20a0c4b33f09309837ffc8bce4c257fc24530","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-16 14:54:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2651cffa3f2d401a8c0dd5eecbbea246\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1c6ae0af8e0e4a6b9ec9db76b92d6a1d : SUCCESS in 2m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2757464f75894b649f217e9600162f90 : SUCCESS in 1m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed42edb341b0489d9c59122c1eedce8c : SUCCESS in 2m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/164fb1d48f3c455c9db04507378571f0 : SUCCESS in 1h 35m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72ac1167653444d88b5e3069b3a155c0 : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7605a454bad5455fbc394ecc1aafc8f6 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6ea737a9c87f43da8ebc477a2023e251 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/077e633e967a4cc6b462020105ba4b41 : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6998651f3c584ede9fba2a0380282b6b : SUCCESS in 2m 56s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/9f1b1304d7724f238bede34e0f9c5a1d : SUCCESS in 16m 35s","accounts_in_message":[],"_revision_number":2},{"id":"caa92c645e431c19a381e745aeceff0638f05a3c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-11-17 08:56:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"733a6ef321fcb1c2c8126833db820ba1aab35fd4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-17 14:00:21.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bf7fa7666e483b65dae84d04d2f8a2cd77c73a34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 14:01:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e9dd69ea37d101871d4bdebc633f555f734d655d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 14:58:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89e32a779858421183b2c04fde7cd4c4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/57ffa78ff14c4b6082e001f9ecd60741 : SUCCESS in 4m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d68b33ce1ef74e44a7c9afb37b12f9ac : SUCCESS in 2m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d6b9fe6278ec4092b62cc29417f7cf24 : SUCCESS in 56m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0f1b5fb733745b0867c75cc79fb00ed : SUCCESS in 2m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b21e6c88d8a54b80aebc364838a40dbb : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/75ef632b2cd14bf19548ac144d0a8d6e : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef7d971868284b8d8bc78867bb630756 : SUCCESS in 3m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53ff96accf4842c5b5c1ce8bc760bd75 : SUCCESS in 3m 40s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/98b76908440b451386d9e868d51461c8 : SUCCESS in 13m 39s","accounts_in_message":[],"_revision_number":2},{"id":"29ef3f980875707dfe85b77a177ca92775776461","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 14:58:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ec7fbf762079616e35be2b40c6e48168cc5a6ae5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 14:59:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a197fb3ae8d4a9e8a75c52330ace9e6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1340fffa7a264388ab726501a3da1531 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa0d388a5e3844f787ad429da7cc6786 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"45104432cb15a79c10532e98240dcc8e2c4b6f3d","revisions":{"d3c2389430e39d394db74ecfcb791d5c3bcdd879":{"kind":"REWORK","_number":1,"created":"2025-11-13 17:28:20.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/71/967071/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/71/967071/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/71/967071/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/71/967071/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/71/967071/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/71/967071/1"}}},"commit":{"parents":[{"commit":"f8e524b83c980ff13026b6098175e118f07eec72","subject":"reno: Update master for unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/f8e524b83c980ff13026b6098175e118f07eec72"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-13 16:00:42.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-13 16:02:46.000000000","tz":0},"subject":"Apply ruff","message":"Apply ruff\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Icb464584eb0abd9014450a63e4b732eda728c3d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/d3c2389430e39d394db74ecfcb791d5c3bcdd879"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/d3c2389430e39d394db74ecfcb791d5c3bcdd879"}]},"branch":"refs/heads/master"},"45104432cb15a79c10532e98240dcc8e2c4b6f3d":{"kind":"REWORK","_number":2,"created":"2025-11-16 13:11:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/71/967071/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/71/967071/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/71/967071/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/71/967071/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/71/967071/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/71/967071/2"}}},"commit":{"parents":[{"commit":"f8e524b83c980ff13026b6098175e118f07eec72","subject":"reno: Update master for unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/f8e524b83c980ff13026b6098175e118f07eec72"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-13 16:00:42.000000000","tz":0},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-16 13:11:16.000000000","tz":0},"subject":"Apply ruff","message":"Apply ruff\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Icb464584eb0abd9014450a63e4b732eda728c3d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/45104432cb15a79c10532e98240dcc8e2c4b6f3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/45104432cb15a79c10532e98240dcc8e2c4b6f3d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
