)]}'
{"id":"openstack%2Foslo.limit~968464","triplet_id":"openstack%2Foslo.limit~master~I21bfa283c95ace092bc154e28f5d584d027c29f9","project":"openstack/oslo.limit","branch":"master","topic":"fixture","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-11-27 06:46:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I21bfa283c95ace092bc154e28f5d584d027c29f9","subject":"Handle missing endpoint, region, service","status":"MERGED","created":"2025-11-26 13:02:18.000000000","updated":"2025-11-27 06:48:14.000000000","submitted":"2025-11-27 06:46:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968464-fixture","meta_rev_id":"6774b9a5035a737a6af7202e933ed9706b1b44a8","_number":968464,"virtual_id_number":968464,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-27 06:46:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-11-27 05:03:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-11-27 05:03:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-26 13:55:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-27 05:03:07.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"b7dc64be29fad413c09ff4626f549d56b85b09dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-26 13:02:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9857fe94a6896b6c376f3a860de79ab9d3992748","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 13:55:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d691c75306c04df1ba18b65041e14719\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9409d65833948d88648f82f4acbbc8b : SUCCESS in 2m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c8673ea34f44a63b5fa122834cdecff : SUCCESS in 51m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1797dee8c47e44408f98b3fb20e0bf9a : SUCCESS in 1m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01aebd0cb402415abeaaaec040f74ca9 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5f33f9bca06d48e281547ecd969d056d : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a06cd2d956b142af81670c636494b75f : SUCCESS in 1m 45s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/d9748390ae8d45de965918ca7e1cbfc5 : SUCCESS in 10m 29s","accounts_in_message":[],"_revision_number":1},{"id":"68b94bed7afff6b4643e8d87364e8e89748e21a1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-27 05:03:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"136b171c4a5a4ea804a3d38acbfd99890e93baad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 05:03:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aed062072a19369d4c0f0ded8106387c2d1826da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 06:46:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e2bb6507b0740eab808f4904f5e7b66\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41db12ead36d40eca389c7fc559a446b : SUCCESS in 54m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b047432069d8431f853c191d80048533 : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/352e940b331e4f4482d6e3a596bb59e4 : SUCCESS in 2m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/35531f64c19b42c69b995791f03e4162 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74c607cad5064a1ca15a67a8dffbacf2 : SUCCESS in 1m 53s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/9dd937149d8746e4b4eb5614a0c378bf : SUCCESS in 21m 39s","accounts_in_message":[],"_revision_number":1},{"id":"de68b03be724f50dcc5be014f18ed265ee703d01","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 06:46:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6774b9a5035a737a6af7202e933ed9706b1b44a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 06:48:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3695fa8e11964ab4a61cf34bc1d9e028\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/197e2138d32d4fb0b8c0b41b7cde7636 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f1ce4022d6d0c7b775dddbf97e1cb0e6194d4dba","revisions":{"f1ce4022d6d0c7b775dddbf97e1cb0e6194d4dba":{"kind":"REWORK","_number":1,"created":"2025-11-26 13:02:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/968464/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/64/968464/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/64/968464/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/64/968464/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/64/968464/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/64/968464/1"}}},"commit":{"parents":[{"commit":"ebfae627732e31ed7eff23afe1bba2eca81a8331","subject":"fixture: Mock out services, regions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/ebfae627732e31ed7eff23afe1bba2eca81a8331"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-26 12:13:50.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-26 13:01:49.000000000","tz":0},"subject":"Handle missing endpoint, region, service","message":"Handle missing endpoint, region, service\n\nSDK will not return \u0027ResourceNotFound\u0027 for a list operation (like\n\u0027services\u0027): it will only do that for a fetch operation (like\n\u0027get_service\u0027):\n\n  \u003e\u003e\u003e import openstack\n  \u003e\u003e\u003e conn \u003d openstack.connect(cloud\u003d\u0027devstack-admin\u0027)\n  \u003e\u003e\u003e set(conn.identity.services(name\u003d\u0027foo\u0027))\n  []\n  \u003e\u003e\u003e conn.identity.get_service(\u0027foo\u0027)\n  Traceback (most recent call last):\n    ...\n  openstack.exceptions.NotFoundException: No Service found for foo: Client Error for url: http://example.com/identity/v3/services/foo, Could not find service: foo.\n\nCorrect our assumptions here.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I21bfa283c95ace092bc154e28f5d584d027c29f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/f1ce4022d6d0c7b775dddbf97e1cb0e6194d4dba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/f1ce4022d6d0c7b775dddbf97e1cb0e6194d4dba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
