)]}'
{"id":"openstack%2Foslo.limit~968473","triplet_id":"openstack%2Foslo.limit~master~Ie878f31d18fbc1e1560349067b2bd8ec448e3227","project":"openstack/oslo.limit","branch":"master","topic":"fixture","attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-11-26 15:00:09.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ie878f31d18fbc1e1560349067b2bd8ec448e3227","subject":"WIP: fixture: Autospec the identity proxy API mock","status":"NEW","created":"2025-11-26 13:24:44.000000000","updated":"2025-11-26 15:00:09.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4ee66f8172a337bcc223da0d5e49c749ac4c9827","_number":968473,"virtual_id_number":968473,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-11-26 15:00:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-26 15:00:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8170320b0eee85553498cea2b83c105c359f1bdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-26 13:24:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ee66f8172a337bcc223da0d5e49c749ac4c9827","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 15:00:09.000000000","message":"Patch Set 1: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/89b07b9c8fe14acb99d3fa23d4588644\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec9afa2461d94c34a39dbf07e8ff87e4 : SUCCESS in 3m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc9d376c9bd94d6eb4025480bde13b78 : SUCCESS in 1h 33m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b07344663c148f3b38b3146401aedf2 : FAILURE in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4fa5f3d463346eabb71bab122a91ee9 : SUCCESS in 6m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5dbced5c5c644d28b61a3de5b2af27e1 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd91a52e6c2c41f3b27ce948f3107e8f : SUCCESS in 2m 25s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/fc5aadcef1ef4ff3b6237f174f5e9b0d : SUCCESS in 24m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0b59fe1bbc4f9ddb806f32060ff27e0c76cb46c6","revisions":{"0b59fe1bbc4f9ddb806f32060ff27e0c76cb46c6":{"kind":"REWORK","_number":1,"created":"2025-11-26 13:24:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/968473/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/73/968473/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/73/968473/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/73/968473/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/73/968473/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/73/968473/1"}}},"commit":{"parents":[{"commit":"f1ce4022d6d0c7b775dddbf97e1cb0e6194d4dba","subject":"Handle missing endpoint, region, service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/f1ce4022d6d0c7b775dddbf97e1cb0e6194d4dba"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-26 13:24:16.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-26 13:24:20.000000000","tz":0},"subject":"WIP: fixture: Autospec the identity proxy API mock","message":"WIP: fixture: Autospec the identity proxy API mock\n\nThis ensures neither we nor other users of the mock can accidentally\ncall or mock a non-existent proxy method.\n\nThis is WIP because the SDK type hints are incorrect and the\ngenerate_fake_proxy proxy method indicates its return type is the\ngeneric Proxy base class instead of a more specific type.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ie878f31d18fbc1e1560349067b2bd8ec448e3227\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/0b59fe1bbc4f9ddb806f32060ff27e0c76cb46c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/0b59fe1bbc4f9ddb806f32060ff27e0c76cb46c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
