)]}'
{"id":"openstack%2Foslo.limit~969407","triplet_id":"openstack%2Foslo.limit~stable%2F2025.2~I21bfa283c95ace092bc154e28f5d584d027c29f9","project":"openstack/oslo.limit","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-12-08 18:11:41.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-12-08 20:02:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I21bfa283c95ace092bc154e28f5d584d027c29f9","subject":"Handle missing endpoint, region, service","status":"MERGED","created":"2025-12-03 12:52:18.000000000","updated":"2025-12-08 20:03:11.000000000","submitted":"2025-12-08 20:02:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969407","meta_rev_id":"24583e6a12ced64d633d407c383b5d4cc6fcded7","_number":969407,"virtual_id_number":969407,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-08 20:02:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-12-08 18:11:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2025-12-08 09:46:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-12-08 18:11:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-03 12:52:18.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2025-12-03 14:26:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-08 09:46:39.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"8ddc7aabca1d8753af7ec582f37b5a36474e891d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-03 12:52:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7afcb82542cbf2ce229fcda94cc3094ae9ed25bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 14:26:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3589dbc37f87492ab0859acbc2642eef\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5637048b19a245a58cae30e75694f2ef : FAILURE in 4m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f321a487e801491ea81a9bef675d91cd : SUCCESS in 1h 28m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee2d58b44b0742d9834b255f97a918fb : FAILURE in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf0c228b3dba400292eee12390ca644b : FAILURE in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5a106fd06bab4032b9f5246d4af62c7b : FAILURE in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c0c431bc65034bf2a523a37b933b7bf2 : FAILURE in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35a00f6603ed4652a390cb41010f11d7 : SUCCESS in 2m 41s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/2c2292858b6f40aeb74c84ed966d8c3f : SUCCESS in 22m 18s\n\nWarning:\n  Comments left for invalid file oslo_limit/fixture.py","accounts_in_message":[],"_revision_number":1},{"id":"d89181e286b6f6eca5de92d101f8955ece829ab8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-03 14:39:21.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"c17582d534053c57dd9f76f55bad7721351d50cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-03 15:26:22.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"381f696104a00dccbea166724309b0917f6cae32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 17:40:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7892f2defbee429d80c60f097545c1e4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5d170e1928a4d09a0b3c53880ba0f34 : SUCCESS in 2m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/71bc00fbc62445ec82be646489c78e9d : SUCCESS in 2h 04m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39115a92592f4934b1892946b21fd765 : FAILURE in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee5438d492494ce698517994bf3ea588 : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9065ffa1f3824d02816de45751b3c56d : SUCCESS in 3m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a57b08342fd54b3d885934376c0c5634 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36d18956d4fd4af2be40a377100d138f : SUCCESS in 4m 23s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/9bde5856d0134cf59dad8c3f2c086b0c : SUCCESS in 25m 34s","accounts_in_message":[],"_revision_number":2},{"id":"43723277e0d450ad663e3d98566fc4af951ab8f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-06 08:22:25.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"432a5eb45947a274274555fd974c94f8f6582d4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-06 09:23:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bcff51f58cdd4f87a27b8758540db3f4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d280b0283f74ed3ba4810ee4b7476ea : SUCCESS in 3m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f8e6648a59cd4e9db248f075ce50f79c : SUCCESS in 1h 00m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9213e5d7bcc4aa0add65f64c2a2f80a : FAILURE in 2m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c855871901af48cc80d7789206af6352 : SUCCESS in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1745c7fb64624489807cbdb3a7e3a2e7 : SUCCESS in 2m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f2a9045c413a48948d9cfed0b67334cf : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/990141eca01549ca8fb42af720bf3945 : SUCCESS in 2m 54s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/dde986ac2cfe4cc1a16ee6eaf8c3b1d6 : SUCCESS in 19m 20s\n\nWarning:\n  Comments left for invalid file oslo_limit/fixture.py","accounts_in_message":[],"_revision_number":3},{"id":"b6c09068e897c6410a290a4459a4ebb377cd5087","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-07 14:41:14.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"1a79b44992e86ba6b3b872107c64dd475c161638","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 15:55:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7600779dac084d21a7fa3652757ad45c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7752b78790ce4b81bc05cd9db6bb6e0a : SUCCESS in 1m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb78c6c2a6c54a5390792203a06e2a16 : SUCCESS in 1h 12m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a55a3463d79e4fa9bc22f894bfb55906 : SUCCESS in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/32d250ca1dc64f3093e891321c1e530a : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a474599452d147d8a14c5c388e09d763 : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/54ec2bfc47464490a89a1fd462088e9a : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdb3310e229a4b3b91d114474099e728 : SUCCESS in 3m 16s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/90b5d85734aa4587a895f1a91924b2b6 : SUCCESS in 10m 21s","accounts_in_message":[],"_revision_number":4},{"id":"d8463261383a403753d5a23fe7f5e4165041d478","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-12-08 09:46:39.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8124b2f364e9e0254c2160a636f702a0393ec53b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-08 18:11:41.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"28e4c982b501fd6681a9b745250ee5a30ea1acf5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-08 18:11:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"19d4fd360b90d96bf8634d7af9efee0ab93b4cf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 18:12:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"cc87b8b49178c084ea3d3ff7a15a6a057fca066d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 20:02:27.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b39071196e714f4a84cad09fdd6ba30c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb36f329a00b42f890997f1f1da30429 : SUCCESS in 1h 45m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc1475d234b8443a85ede3ac0ac9f66d : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c019e55015e94c21a1f10ca9a3cbfffd : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b0f46e995be74f3d958d830cb2443c9a : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0df683f28fc413f97ff65510973f7f6 : SUCCESS in 3m 03s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/49b2812b6ecc43f7b470ca0794a6400a : SUCCESS in 10m 17s","accounts_in_message":[],"_revision_number":4},{"id":"609fb9ab97728c6938d15b2d149f7f0f74c11b08","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 20:02:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"24583e6a12ced64d633d407c383b5d4cc6fcded7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 20:03:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d05e1a4ea59a4d89a64f7edfeba07c67\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4df51f6c0674b278521fc5c169e74d4 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"20e56c8e6d52c62c898288aee2dfaeb25871edb9","revisions":{"589f3b3193f3116523c61b8cadd2152e81a804e4":{"kind":"REWORK","_number":1,"created":"2025-12-03 12:52:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/969407/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/07/969407/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/07/969407/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/07/969407/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/07/969407/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/07/969407/1"}}},"commit":{"parents":[{"commit":"d0d891010eb33b1ee93a6ed538c8ace2e42dc6f6","subject":"fixture: Mock out services, regions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/d0d891010eb33b1ee93a6ed538c8ace2e42dc6f6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-26 12:13:50.000000000","tz":0},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-03 12:52:09.000000000","tz":540},"subject":"Handle missing endpoint, region, service","message":"Handle missing endpoint, region, service\n\nSDK will not return \u0027ResourceNotFound\u0027 for a list operation (like\n\u0027services\u0027): it will only do that for a fetch operation (like\n\u0027get_service\u0027):\n\n  \u003e\u003e\u003e import openstack\n  \u003e\u003e\u003e conn \u003d openstack.connect(cloud\u003d\u0027devstack-admin\u0027)\n  \u003e\u003e\u003e set(conn.identity.services(name\u003d\u0027foo\u0027))\n  []\n  \u003e\u003e\u003e conn.identity.get_service(\u0027foo\u0027)\n  Traceback (most recent call last):\n    ...\n  openstack.exceptions.NotFoundException: No Service found for foo: Client Error for url: http://example.com/identity/v3/services/foo, Could not find service: foo.\n\nCorrect our assumptions here.\n\nConflicts:\n        oslo_limit/limit.py\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I21bfa283c95ace092bc154e28f5d584d027c29f9\n(cherry picked from commit f1ce4022d6d0c7b775dddbf97e1cb0e6194d4dba)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/589f3b3193f3116523c61b8cadd2152e81a804e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/589f3b3193f3116523c61b8cadd2152e81a804e4"}]},"branch":"refs/heads/stable/2025.2"},"e302135be66e637321783d1e2a6e520d8f99a27f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-12-03 15:26:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/969407/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/07/969407/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/07/969407/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/07/969407/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/07/969407/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/07/969407/2"}}},"commit":{"parents":[{"commit":"5762227e49d5ea0e709889acbabc56bd40b1eae6","subject":"fixture: Mock out services, regions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/5762227e49d5ea0e709889acbabc56bd40b1eae6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-26 12:13:50.000000000","tz":0},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-03 15:26:22.000000000","tz":0},"subject":"Handle missing endpoint, region, service","message":"Handle missing endpoint, region, service\n\nSDK will not return \u0027ResourceNotFound\u0027 for a list operation (like\n\u0027services\u0027): it will only do that for a fetch operation (like\n\u0027get_service\u0027):\n\n  \u003e\u003e\u003e import openstack\n  \u003e\u003e\u003e conn \u003d openstack.connect(cloud\u003d\u0027devstack-admin\u0027)\n  \u003e\u003e\u003e set(conn.identity.services(name\u003d\u0027foo\u0027))\n  []\n  \u003e\u003e\u003e conn.identity.get_service(\u0027foo\u0027)\n  Traceback (most recent call last):\n    ...\n  openstack.exceptions.NotFoundException: No Service found for foo: Client Error for url: http://example.com/identity/v3/services/foo, Could not find service: foo.\n\nCorrect our assumptions here.\n\nConflicts:\n        oslo_limit/limit.py\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I21bfa283c95ace092bc154e28f5d584d027c29f9\n(cherry picked from commit f1ce4022d6d0c7b775dddbf97e1cb0e6194d4dba)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/e302135be66e637321783d1e2a6e520d8f99a27f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/e302135be66e637321783d1e2a6e520d8f99a27f"}]},"branch":"refs/heads/stable/2025.2","description":"Rebase"},"9910844956ff5a48347bbe7a67737906255075d5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-12-06 08:22:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/969407/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/07/969407/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/07/969407/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/07/969407/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/07/969407/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/07/969407/3"}}},"commit":{"parents":[{"commit":"9994657ed83e3724e152ba5ba811c224ae4c5784","subject":"fixture: Mock out services, regions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/9994657ed83e3724e152ba5ba811c224ae4c5784"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-26 12:13:50.000000000","tz":0},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-06 08:22:25.000000000","tz":0},"subject":"Handle missing endpoint, region, service","message":"Handle missing endpoint, region, service\n\nSDK will not return \u0027ResourceNotFound\u0027 for a list operation (like\n\u0027services\u0027): it will only do that for a fetch operation (like\n\u0027get_service\u0027):\n\n  \u003e\u003e\u003e import openstack\n  \u003e\u003e\u003e conn \u003d openstack.connect(cloud\u003d\u0027devstack-admin\u0027)\n  \u003e\u003e\u003e set(conn.identity.services(name\u003d\u0027foo\u0027))\n  []\n  \u003e\u003e\u003e conn.identity.get_service(\u0027foo\u0027)\n  Traceback (most recent call last):\n    ...\n  openstack.exceptions.NotFoundException: No Service found for foo: Client Error for url: http://example.com/identity/v3/services/foo, Could not find service: foo.\n\nCorrect our assumptions here.\n\nConflicts:\n        oslo_limit/limit.py\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I21bfa283c95ace092bc154e28f5d584d027c29f9\n(cherry picked from commit f1ce4022d6d0c7b775dddbf97e1cb0e6194d4dba)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/9910844956ff5a48347bbe7a67737906255075d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/9910844956ff5a48347bbe7a67737906255075d5"}]},"branch":"refs/heads/stable/2025.2","description":"Rebase"},"20e56c8e6d52c62c898288aee2dfaeb25871edb9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-12-07 14:41:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/969407/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/07/969407/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/07/969407/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/07/969407/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/07/969407/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/07/969407/4"}}},"commit":{"parents":[{"commit":"345fb4cf94876be522185126e9fec28e65db1bf5","subject":"fixture: Mock out services, regions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/345fb4cf94876be522185126e9fec28e65db1bf5"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-26 12:13:50.000000000","tz":0},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 14:41:14.000000000","tz":0},"subject":"Handle missing endpoint, region, service","message":"Handle missing endpoint, region, service\n\nSDK will not return \u0027ResourceNotFound\u0027 for a list operation (like\n\u0027services\u0027): it will only do that for a fetch operation (like\n\u0027get_service\u0027):\n\n  \u003e\u003e\u003e import openstack\n  \u003e\u003e\u003e conn \u003d openstack.connect(cloud\u003d\u0027devstack-admin\u0027)\n  \u003e\u003e\u003e set(conn.identity.services(name\u003d\u0027foo\u0027))\n  []\n  \u003e\u003e\u003e conn.identity.get_service(\u0027foo\u0027)\n  Traceback (most recent call last):\n    ...\n  openstack.exceptions.NotFoundException: No Service found for foo: Client Error for url: http://example.com/identity/v3/services/foo, Could not find service: foo.\n\nCorrect our assumptions here.\n\nConflicts:\n        oslo_limit/limit.py\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I21bfa283c95ace092bc154e28f5d584d027c29f9\n(cherry picked from commit f1ce4022d6d0c7b775dddbf97e1cb0e6194d4dba)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/20e56c8e6d52c62c898288aee2dfaeb25871edb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/20e56c8e6d52c62c898288aee2dfaeb25871edb9"}]},"branch":"refs/heads/stable/2025.2","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
