)]}'
{"id":"openstack%2Foslo.limit~986558","triplet_id":"openstack%2Foslo.limit~master~Ib522130ed6bbceda2f408e8fae3b5dc4437ea2d2","project":"openstack/oslo.limit","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-04-29 10:28:50.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-04-29 11:55:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib522130ed6bbceda2f408e8fae3b5dc4437ea2d2","subject":"Fix new mypy errors","status":"MERGED","created":"2026-04-28 16:24:03.000000000","updated":"2026-04-29 11:56:37.000000000","submitted":"2026-04-29 11:55:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986558","meta_rev_id":"5d0820283f57d33943775466396205559b2fc0c2","_number":986558,"virtual_id_number":986558,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-29 11:55:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-04-29 10:28:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-04-29 10:28:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-28 16:29:04.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2026-04-28 19:01:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-29 10:28:50.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"4c411724e4c6ee3c244492e49c065bd9b0b1639d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-28 16:24:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd7b535d274bc7744e283ab7523fe009376c8ab8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-28 16:29:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"358089641c8d20fd207298a8e60bbfa4f8e44ef6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-28 16:34:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"27997b4c5c75a59354e15da640e10a3feaf25094","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-28 16:34:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"edea55a7c921b1e195070e3c3cabeba1b5c1ad53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-28 16:48:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d88c06f009c2f1ff64bc7cdea4edeeb99a68a103","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 19:01:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2861d1b99a6043208b818d8539449897\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8517d9e6b1ac44c58af629dd9fcbf195 : SUCCESS in 2m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dad4d831a2df4445a84e4a9364f2b058 : SUCCESS in 3m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8f56a12252c4db5b7287416b2618012 : SUCCESS in 2h 07m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b76585a0abbb4d2dae2e6b28327f208b : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2995fa3d617a42a4843ee18d969eb3f0 : SUCCESS in 2m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/293e62a5350645d9a44187b617ffc666 : SUCCESS in 4m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/473b021105e446f68e3d5cb34afa9367 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe0c56e6816442178bb435225243c859 : SUCCESS in 3m 21s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/b2e042f9142148e5887263e8bbb26a7c : SUCCESS in 33m 18s","accounts_in_message":[],"_revision_number":3},{"id":"09b4630d083307566a23691860e376ab5a809b8c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-29 10:28:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"824d5ae420eddd5b4ce1f25955d287c5a381d5db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 10:29:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b923f7e7d3c5ecaf2272f4a411a1735c6f717c27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 11:55:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c430569e2e334b40a7155c4e11abf86c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b42af23418e641b4b4c9e3fcdad7a647 : SUCCESS in 1m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/47ac28e72385468ebba2e53d445efdc2 : SUCCESS in 1h 24m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a377aee66254a32bd9babcc00095698 : SUCCESS in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/540c835a4e624eab9ce87dd8d0614a61 : SUCCESS in 2m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6006d790f1f449da910ca77c822c17ff : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73778454ee5a42d090655b629d380e76 : SUCCESS in 3m 27s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/1b4c16aca6c1486da1f844287d4d1395 : SUCCESS in 15m 20s","accounts_in_message":[],"_revision_number":3},{"id":"7b5228e32205ad46a667d271dc1d88c1398f1ad0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 11:55:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5d0820283f57d33943775466396205559b2fc0c2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 11:56:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf17fcede9cd42fba517bb7d1294d102\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71dcfb0eb20a4eca89e293e2e1ff5641 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"01cfe71c5653c87ff579c296dc48133cb5c6c9a2","revisions":{"179d43bbaf5b4006688b83fb669f001c2ecc8f3a":{"kind":"REWORK","_number":1,"created":"2026-04-28 16:24:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/986558/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/58/986558/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/58/986558/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/58/986558/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/58/986558/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/58/986558/1"}}},"commit":{"parents":[{"commit":"8cc6b0a0e18531af98006d590e08d20628abe444","subject":"ruff: Configure hacking as external linter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/8cc6b0a0e18531af98006d590e08d20628abe444"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-28 16:23:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-28 16:23:56.000000000","tz":540},"subject":"Fix new mypy errors","message":"Fix new mypy errors\n\nFix the new errors detected by additional type hints from openstacksdk.\n\nChange-Id: Ib522130ed6bbceda2f408e8fae3b5dc4437ea2d2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/179d43bbaf5b4006688b83fb669f001c2ecc8f3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/179d43bbaf5b4006688b83fb669f001c2ecc8f3a"}]},"branch":"refs/heads/master"},"3641e810726ef6969910405686dd58211a4c4866":{"kind":"REWORK","_number":2,"created":"2026-04-28 16:34:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/986558/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/58/986558/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/58/986558/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/58/986558/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/58/986558/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/58/986558/2"}}},"commit":{"parents":[{"commit":"8cc6b0a0e18531af98006d590e08d20628abe444","subject":"ruff: Configure hacking as external linter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/8cc6b0a0e18531af98006d590e08d20628abe444"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-28 16:23:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-28 16:34:23.000000000","tz":540},"subject":"Fix new mypy errors","message":"Fix new mypy errors\n\nFix the new errors detected by additional type hints from openstacksdk.\n\nChange-Id: Ib522130ed6bbceda2f408e8fae3b5dc4437ea2d2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3641e810726ef6969910405686dd58211a4c4866"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3641e810726ef6969910405686dd58211a4c4866"}]},"branch":"refs/heads/master"},"01cfe71c5653c87ff579c296dc48133cb5c6c9a2":{"kind":"REWORK","_number":3,"created":"2026-04-28 16:48:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/986558/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/58/986558/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/58/986558/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/58/986558/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/58/986558/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/58/986558/3"}}},"commit":{"parents":[{"commit":"8cc6b0a0e18531af98006d590e08d20628abe444","subject":"ruff: Configure hacking as external linter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/8cc6b0a0e18531af98006d590e08d20628abe444"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-28 16:23:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-28 16:47:49.000000000","tz":540},"subject":"Fix new mypy errors","message":"Fix new mypy errors\n\nFix the new errors detected by additional type hints from openstacksdk.\n\nChange-Id: Ib522130ed6bbceda2f408e8fae3b5dc4437ea2d2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/01cfe71c5653c87ff579c296dc48133cb5c6c9a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/01cfe71c5653c87ff579c296dc48133cb5c6c9a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
