)]}'
{"id":"openstack%2Foslo.limit~992644","triplet_id":"openstack%2Foslo.limit~master~Ic88e72ae77b0fa5551f897a2ce240bbef36706dc","project":"openstack/oslo.limit","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-06-22 12:23:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic88e72ae77b0fa5551f897a2ce240bbef36706dc","subject":"Remove Python 3.10 support","status":"MERGED","created":"2026-06-10 13:32:20.000000000","updated":"2026-06-22 12:24:48.000000000","submitted":"2026-06-22 12:23:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"992644","meta_rev_id":"7fbec5a9a1f185bc1000dff469343e17c7724d56","_number":992644,"virtual_id_number":992644,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-22 12:23:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-06-10 13:52:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-06-10 13:52:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-10 13:52:49.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-06-10 15:40:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cf8163eecff02ea2b7bfad90c13d471459592666","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-10 13:32:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"485b73835222ceead667a3eb821d9a2f169479a2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-10 13:52:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2cf3fb9b928eb20d460700a820efdc4d82890957","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 15:40:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/092377003e9f475a9f0cc602d9d28d9a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5d0bc8ca9e794c72b5d5032926d0b116 : SUCCESS in 3m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1dfaf7e3149743ceb034fe8e17db16b5 : SUCCESS in 4m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1481e90493ac409f9572536b117ac130 : SUCCESS in 4m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/247be740e59042e2a26fff90290ec2d7 : SUCCESS in 1h 58m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/186f4c504a644012a2ae6bf793724185 : FAILURE in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac12f207baba4204af8e09090c0222ba : SUCCESS in 3m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5547f5aa366f44aaa094e924dfcdf3dd : SUCCESS in 4m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/da1cca963a704900a34924995cfa9b82 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1602446edf541c0912e17d470a6205e : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6e133b3927d494d8877acf163521fd7 : SUCCESS in 3m 51s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/7313cc7e91644863ab41c2258ed3f6ae : SUCCESS in 29m 41s\n\nWarning:\n  Comments left for invalid file oslo_limit/limit.py","accounts_in_message":[],"_revision_number":1},{"id":"4438b67ee9e42000ff7c77434a6579850faa3a19","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-22 09:27:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"53a4b0aaeab91157d721dd8a5713f88ed6238514","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 11:22:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffdd3edb366c496192854866b316b613\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a5bab228935e4b6da162d032efb18563 : SUCCESS in 2m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f52e8d5ecf264a74b55f51043fe999ec : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/602973ecc6a544e08ced1389204ab056 : SUCCESS in 3m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3ccc5d86cb042cfa4ebe5b3d756230a : SUCCESS in 1h 41m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14a87faf043f4112a963067ba295fcb1 : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc6d572e3d1a47fc82141267f7978f05 : SUCCESS in 2m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1cda9e1f6e304d8f8e174e3ebabd2137 : SUCCESS in 3m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c882d47e9e994377bf8f7e6bfd84e74d : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f805eab6c3514505a8261a70b647ceb3 : SUCCESS in 3m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d99db119d744a29bb15cd0ea1736c62 : SUCCESS in 3m 40s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/7e07740fe51f47ffa7e047ef8fb5d3be : SUCCESS in 25m 28s","accounts_in_message":[],"_revision_number":1},{"id":"77276ab1a48853b382f0aad71a3ffb312e6c4fa3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 11:23:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee7c3467fd790825dd7a6c61334adf3bae61417c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 12:23:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffa63b29c178482e985598804916421d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/42c420724c4c4fb5ac9a6bb1142a491e : SUCCESS in 2m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7197009815c24f8c80daa2a1bed41ceb : SUCCESS in 2m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/84dbacd3631f4a0893eb907523f5c4f5 : SUCCESS in 59m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d7a7447ad6241f99f5024d026f331d0 : SUCCESS in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ccecdfd51dfb44d8bf4949b1bd23b631 : SUCCESS in 2m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3c9a0fe9d81c45e3a66a17e73e1feb7b : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df73bae98f6e4047b7f8c5eef2be58d4 : SUCCESS in 2m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0493d191860e4903a6b42d0b03bf2248 : SUCCESS in 2m 54s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/066456202dd847a891b0adb3c782bad4 : SUCCESS in 19m 16s","accounts_in_message":[],"_revision_number":1},{"id":"48a7f9648bfd7dcf6414cf554401f6c01bcb43ea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 12:23:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7fbec5a9a1f185bc1000dff469343e17c7724d56","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 12:24:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c3eef0d097e4780afa1deb7ff2ec946\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96212c336e0f4e3ab6c0fd7ca869564a : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78d6f21a7aec42479070ec2ab3fff7e3 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9e139273e2cbf7da7292bed3f1c825282dc7a541","revisions":{"9e139273e2cbf7da7292bed3f1c825282dc7a541":{"kind":"REWORK","_number":1,"created":"2026-06-10 13:32:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/992644/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/44/992644/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/44/992644/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/44/992644/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/44/992644/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/44/992644/1"}}},"commit":{"parents":[{"commit":"697a4fc23cc6003ce885569225848374c42ba605","subject":"Merge \"Update packaging configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/697a4fc23cc6003ce885569225848374c42ba605"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-10 13:32:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-10 13:32:08.000000000","tz":540},"subject":"Remove Python 3.10 support","message":"Remove Python 3.10 support\n\nPython 3.10 is no longer part of the tested runtimes.\n\nChange-Id: Ic88e72ae77b0fa5551f897a2ce240bbef36706dc\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/9e139273e2cbf7da7292bed3f1c825282dc7a541"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/9e139273e2cbf7da7292bed3f1c825282dc7a541"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
