)]}'
{"id":"openstack%2Foslo.log~1000186","triplet_id":"openstack%2Foslo.log~master~If5819dac4fd6c47965b1edbd84cee50ced0988ee","project":"openstack/oslo.log","branch":"master","topic":"release-health-check","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-08-27 11:09:10.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If5819dac4fd6c47965b1edbd84cee50ced0988ee","subject":"DNM: gate health test","status":"ABANDONED","created":"2026-08-08 09:42:31.000000000","updated":"2026-08-27 11:09:10.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fe164148fbe44fc247953353ef5090f7552cf27a","_number":1000186,"virtual_id_number":1000186,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-27 10:12:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-08 11:22:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c767a29db43c709b95c32cae7b4ac9cba03abf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-08-08 09:42:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f874ae3d048762842b42358c6093bc375fa8428c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-08 11:22:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a36b509aa614d908e43ad3889e1bb47\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d31ff7f58e114a9a86b8a0067d945ec6 : SUCCESS in 1h 34m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31118a7ebc244ed9b9faf415f062851c : SUCCESS in 3m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5fe3b7eebeb14a2080a27c415b133406 : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a4732e2256ad4f8ab90e13fc412e2485 : SUCCESS in 3m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/69ac53482c894f9ba5f0051559c4c929 : FAILURE in 7m 11s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/dff0dac4de674c31b7e0d951b656e2f4 : FAILURE in 8m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57ddc9fd614445a390349a8532e057e3 : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a46fcc84910745d08b64bc05bce7e5f8 : SUCCESS in 2m 27s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/f8ce65fa73054584b1b40184b54e942a : SUCCESS in 1h 03m 42s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/ca79fca8f38249e890873fed784f9539 : SUCCESS in 32m 32s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/d8fb66dc2e07400aa4277856a31ef762 : SUCCESS in 3m 44s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/f5574f9861b6498ca56c1e03fa3a4395 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":1},{"id":"7846f013db8d7486bccb6445db46d131fb753e0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"real_author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-23 22:10:19.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_17685\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"_revision_number":2},{"id":"f33fceaf3fa380512f26e4bbc3773c6822ecdee1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-23 22:59:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/935f225df4134b4da302e6ea9aef5796\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3c74fbb7e35041729dc70785f297e4ff : SUCCESS in 48m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53e18d903f674b4186611653eb51fa57 : FAILURE in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9e4476becea34fe7901a56dd32fda646 : SUCCESS in 2m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b5b1ac45d3c146c98b026ace7350d414 : SUCCESS in 2m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/581a46a5734b41159685b2b50546065e : SUCCESS in 3m 50s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/3e3d07bd354647a1a8e3b96606077ec3 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ab74eba7a4e4d59b11c0393a8e7c8d7 : SUCCESS in 2m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c64011a28448446f9302c1aea15bed87 : SUCCESS in 2m 18s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/b98b0727cf32441faee00ba3cbb41faa : SUCCESS in 30m 04s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/4d330195fa484a709881a172bb1ba53c : SUCCESS in 17m 13s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/305d48db8f5b43ad8d69394f0354c055 : SUCCESS in 2m 02s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/dab9a12f6129408fb7f324fc7c7d9495 : SUCCESS in 1m 56s\n\nWarning:\n  Comments left for invalid file oslo_log/versionutils.py","accounts_in_message":[],"_revision_number":2},{"id":"ddaab1a17f5228342f259f2257a2fe40044c54b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-08-27 09:05:29.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"b82f5e7983062f253eba42f04147602de3900e90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 10:12:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb961b585ac148a8a08d56521caf705b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b40409f83e0b4412a6e6cb1e4726c6a7 : SUCCESS in 45m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71be6add2d7c433a98adffc6049d7003 : SUCCESS in 6m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/69ce203401894120a400a52fe6b2eb97 : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/34edec3095aa4e698c3e2a0a54c51037 : SUCCESS in 2m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/31607c61ccd8408aafd1f045f4c4fe4c : SUCCESS in 7m 26s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/696eb83e19814047a1d05e5e9d69565e : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6bc2d5d2e95408db9f8178d893e0240 : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0c9d90a338b4c3dbcb7efb6e851add4 : SUCCESS in 3m 05s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/34a427cd396e4efaaa641942056447eb : SUCCESS in 1h 04m 18s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/43c2b7e2153a46b3ab7b7e8cbc5f0dd5 : SUCCESS in 38m 46s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/2e58d7ed961740559d86d389e79eef3f : SUCCESS in 3m 36s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/aaf2a0565dcc4e2697b2d769b99cd31e : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":3},{"id":"fe164148fbe44fc247953353ef5090f7552cf27a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-27 11:09:10.000000000","message":"Abandoned\n\nLooks like this back working as expected","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"eeb93ac272d6c5c68843ef38b241f76344e07d75","revisions":{"600f899e92c9734c63d9703ba7368211282b9c38":{"kind":"REWORK","_number":1,"created":"2026-08-08 09:42:31.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/86/1000186/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/86/1000186/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/1000186/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/1000186/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/1000186/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/86/1000186/1"}}},"commit":{"parents":[{"commit":"7808001c105595d3e2bd856afba1fc57653aa711","subject":"zuul: Use openstack-python3-next-jobs template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/7808001c105595d3e2bd856afba1fc57653aa711"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-08-08 09:42:21.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-08-08 09:42:29.000000000","tz":120},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: If5819dac4fd6c47965b1edbd84cee50ced0988ee\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/releases:tools/repository_test_generator.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/600f899e92c9734c63d9703ba7368211282b9c38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/600f899e92c9734c63d9703ba7368211282b9c38"}]},"branch":"refs/heads/master"},"6f07df569c1c3f672c2ffed78a4d7c2f8912117f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-08-23 22:10:19.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"real_uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/86/1000186/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/86/1000186/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/1000186/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/1000186/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/1000186/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/86/1000186/2"}}},"commit":{"parents":[{"commit":"6c5e0d418038c96d071d36a6ca9f20262e509df9","subject":"Address Python 3.13 issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/6c5e0d418038c96d071d36a6ca9f20262e509df9"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-08-08 09:42:21.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-08-23 22:10:19.000000000","tz":0},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: If5819dac4fd6c47965b1edbd84cee50ced0988ee\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/releases:tools/repository_test_generator.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/6f07df569c1c3f672c2ffed78a4d7c2f8912117f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/6f07df569c1c3f672c2ffed78a4d7c2f8912117f"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"7808001c105595d3e2bd856afba1fc57653aa711","ours":"600f899e92c9734c63d9703ba7368211282b9c38","theirs":"6c5e0d418038c96d071d36a6ca9f20262e509df9","merge_strategy":"recursive","contains_conflicts":false}},"eeb93ac272d6c5c68843ef38b241f76344e07d75":{"kind":"REWORK","_number":3,"created":"2026-08-27 09:05:29.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/86/1000186/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/86/1000186/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/1000186/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/1000186/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/1000186/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/86/1000186/3"}}},"commit":{"parents":[{"commit":"6c5e0d418038c96d071d36a6ca9f20262e509df9","subject":"Address Python 3.13 issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/6c5e0d418038c96d071d36a6ca9f20262e509df9"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-08-08 09:42:21.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-08-27 09:05:29.000000000","tz":0},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: If5819dac4fd6c47965b1edbd84cee50ced0988ee\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/releases:tools/repository_test_generator.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/eeb93ac272d6c5c68843ef38b241f76344e07d75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/eeb93ac272d6c5c68843ef38b241f76344e07d75"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
