)]}'
{"id":"openstack%2Foslo.log~1004696","triplet_id":"openstack%2Foslo.log~stable%2F2026.1~I24a59b51d361e26bca6e5a6d54b75afaca01d5a8","project":"openstack/oslo.log","branch":"stable/2026.1","topic":"fix_deprecated_list","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-09-09 18:44:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I24a59b51d361e26bca6e5a6d54b75afaca01d5a8","subject":"Fix the ``deprecated`` version names","status":"MERGED","created":"2026-09-08 20:40:54.000000000","updated":"2026-09-09 18:45:51.000000000","submitted":"2026-09-09 18:44:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004696-fix_deprecated_list","meta_rev_id":"1ef501137fa7ec0b511031d9674817836b1966ed","_number":1004696,"virtual_id_number":1004696,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-09 18:44:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-09 10:20:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-09-09 14:48:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2026-09-09 14:48:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-08 22:24:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-09 10:20:53.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-09-09 14:48:52.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"30649269a1ac697003fc9afe15c57daf7251dcee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-08 20:40:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e1e20cb5da954a37826f03bde413306cd808d7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 22:24:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f17cb5139c941ffac16b02b77cd014f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/04dd36c084844d11a1627b1f50b68efd : SUCCESS in 1h 37m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2c52a32d99f4d9ebf849bcf8ba751b7 : FAILURE in 32m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/536f923e436c4058b12b17bd05866636 : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f048b6db51424b16a54b52ac0cdd1b95 : SUCCESS in 4m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d69caed757a3442bbaa2af16a33ab0c6 : FAILURE in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bd16d85c1a540a4ade229625b1959cc : SUCCESS in 2m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2497be975194186bfe22fa88c94e68b : SUCCESS in 3m 21s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/e1f327d6922748e2be15bd83007985f0 : SUCCESS in 1h 14m 39s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/059d2c8bf2ba419687e3b143ea1622a6 : SUCCESS in 25m 27s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/930f928058b14768a025fb960ce7b732 : SUCCESS in 3m 37s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/c530a065ce3344c58e76a6cb9029941e : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"86d31793e05f916e66f88fd8541c129cfc265005","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-09 10:20:53.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"71f9bcd8b90ddd2b7f244165aeda2bfe0ecbe116","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 12:09:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a743c7a41d154d48a2a36621de1001b8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/23d1bb5294af49c4934acf6b29caba86 : SUCCESS in 1h 41m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5136992f69904f2caf4e1158a9380c6e : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b057d5ed62574955aa0c832a745e0ec0 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c4092174bed9486892b61418b1281324 : SUCCESS in 4m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5befb1f4bb5e4fae82de91bfd1e0e184 : FAILURE in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6128997540be4850aa83dd6bbb88602e : POST_FAILURE in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1989244e2d0e467f8e55b39b678a9abd : SUCCESS in 3m 22s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/5511e42d5b644072ba8ee7525585b92a : SUCCESS in 44m 47s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/bb116d7c5944414cbbedc5eebb1110b1 : SUCCESS in 23m 11s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/2acedbec528f465e9a2dd9bde80efbcb : SUCCESS in 3m 29s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/e8b9d70840c3433db692a350829e3e13 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":1},{"id":"4933096899b53dab5902f6d9601c8395153def42","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-09 14:48:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"130255606a5aad1d002eb495c2cf4f1a3fdbbe1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 16:46:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d5e048a2c224a9097393a6c9de10a35\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d463f168297445d81acdeac5f296e47 : SUCCESS in 1h 44m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59b256b9a1c04a4fa9e31b249b7308a0 : SUCCESS in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8627f0327f74e1aa7c00a78860855d2 : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/13306b933a944e9190e30d884801f258 : SUCCESS in 3m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/771e4c078127483995bdf3ca4d85f8e8 : FAILURE in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62b5787ca71f4efea592cd2afc9fe2c4 : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf7bd77e41e34453b47e1d36efdd84e3 : SUCCESS in 3m 24s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/67785f633bca44ff8182c49bbf4847f1 : SUCCESS in 38m 15s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/3c644b57d70b4c7ca0138493fe945ec3 : SUCCESS in 38m 38s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/1050cb53fa3e4f49a40e92c5b4227a5b : SUCCESS in 3m 37s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/be063023b38f49758c1e5004e055c07a : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":1},{"id":"4c849e8460ea7a2eac295b2c4d0f6fa5d9598d57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 16:47:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"35aba99346a58d90f65bbcf20368cb84f02fa8ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 18:44:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06154f68acb542388c71dc6e16e7f8f9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/98103e74c0494476b30d055a4b4aeab6 : SUCCESS in 1h 49m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8141d7884b9e4cb7ace1936d24f08b62 : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1abe3478fac4eec9035f7db8a6632e3 : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ae50bcb78915473ab2beec359dafd1e4 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f317d3738452447fbef3a1efb5fa205a : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b357b4649ff34a32b7e762f6a35d093d : SUCCESS in 3m 42s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/dc5af9976e844febb277ac2b339b8243 : SUCCESS in 26m 21s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/0cf01ea55d41482db47053a5bd31577a : SUCCESS in 3m 13s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/c44da7edae714efebac30289a276b7ec : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":1},{"id":"53d4650a23b8fd400c8d23b7bd8dddab032b764b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 18:44:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1ef501137fa7ec0b511031d9674817836b1966ed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 18:45:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b49b67d74734bebb9ab33db8808eb7b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20ae89eaf7524eb0a1b858fdbdbd07ad : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b715fc44327f9f3cef714cd9301de32355ca3274","revisions":{"b715fc44327f9f3cef714cd9301de32355ca3274":{"kind":"REWORK","_number":1,"created":"2026-09-08 20:40:54.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/96/1004696/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/96/1004696/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/96/1004696/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/96/1004696/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/96/1004696/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/96/1004696/1"}}},"commit":{"parents":[{"commit":"14f9dd993ca1a2e7444a41881dce2c19a68642ee","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/14f9dd993ca1a2e7444a41881dce2c19a68642ee"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-09-08 13:45:44.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-09-08 20:40:48.000000000","tz":120},"subject":"Fix the ``deprecated`` version names","message":"Fix the ``deprecated`` version names\n\n``FLAMINGO`` and ``GAZPACHO`` are pointing to wrong version numbers.\n2025.2 is missing in this list.\n\nThis patch adds INDRI \u003d \u00272027.1\u0027 to this list.\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I24a59b51d361e26bca6e5a6d54b75afaca01d5a8\n(cherry picked from commit 840d6e2c996e15556a55ca86da1bd40b2a1470ae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/b715fc44327f9f3cef714cd9301de32355ca3274"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/b715fc44327f9f3cef714cd9301de32355ca3274"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
