)]}'
{"id":"openstack%2Foslo.log~1005498","triplet_id":"openstack%2Foslo.log~master~I96ed43eee662408ed6b26800afae74de3963c7e1","project":"openstack/oslo.log","branch":"master","hashtags":[],"change_id":"I96ed43eee662408ed6b26800afae74de3963c7e1","subject":"Ensure recursive detection of package files","status":"MERGED","created":"2026-09-14 05:09:46.000000000","updated":"2026-09-14 08:37:41.000000000","submitted":"2026-09-14 08:33:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005498","meta_rev_id":"ac1d8d6644af021a0bacc03336a8e03651614b3c","_number":1005498,"virtual_id_number":1005498,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-14 08:33:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-09-14 07:39:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-09-14 07:39:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-14 06:30:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62655979101da3e067d8f85c9a8d5fb0fc29e53f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-14 05:09:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b204bbdb5a6e50fa5d9f8944800d3356a698be91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 06:30:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57a0eec9121042b185e237fcb8c15f64\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b012b7f6bf0b4b3caae7097d275ac16c : SUCCESS in 2m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/909902e181214dc7b1417edbd0ac7aa1 : SUCCESS in 2m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/94293d0465c041679c4b41dafc9adc3a : SUCCESS in 1h 19m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e03eb21a3414a9aa0428e2d02bfa98c : SUCCESS in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ead1290c1f304e1c8098962c68f596de : SUCCESS in 2m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dde1429840aa44429f8b9acd9b917434 : SUCCESS in 3m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/defeb427a1eb40758607ee2fd71ed0f9 : SUCCESS in 7m 17s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/0e9031e430e04c01b1ea60196d75678f : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bd0c3207d544192b2ac2ba0b8435ac9 : SUCCESS in 2m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afa21ccbe6b240998d754b75aa9b7202 : SUCCESS in 3m 08s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/df55a82fdfab4c25bfa578345b175be3 : SUCCESS in 1h 05m 24s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/0feceb1bbe49429c8c3891dd26de5484 : SUCCESS in 38m 04s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/1ee7b4ac7c1b4cd590b9fc25749fa3b7 : SUCCESS in 2m 33s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/e4ba50a0b4584afe9d5be1708c5c98e3 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"3e14faa47a1a3843aac55d5b1078137e978debeb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-14 07:39:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"470376c15ea19f31c8e15198afb71a8d3657f16e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 07:41:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1b0c46fa91b3a5e6c526949c42f3347c7562e214","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 08:33:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da27b9e6ac0d4c7b800702404fe9d3fb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0e7352b1d93f4741aa26a526859e6dd1 : SUCCESS in 2m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2ec98f37920d4262a8234c4537fbbdbc : SUCCESS in 2m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fab303afa61f479ebed91ded9eb31b99 : SUCCESS in 52m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/377495f80f154b13a7936ff7de92c37c : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a3995c10d5c94526add2b13f554f2c7f : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/73466295b89e40e69b47374461c81386 : SUCCESS in 3m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/423ce797194a4acaa792233cfe480f8b : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff07ea9b27114b27a77507425e5fc55d : SUCCESS in 2m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b4dff6054e043d1a71a5fc4eab4d2d4 : SUCCESS in 3m 23s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/9877498ee75c433698c3a8a62f5247ac : SUCCESS in 17m 45s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/1f283103feed4e9280e107f9abaeda3b : SUCCESS in 3m 12s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/1297871d6d384271be87d358380ca295 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"3f9a723f5fae0d9c0f2598595440fcf7a94e5997","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 08:33:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ac1d8d6644af021a0bacc03336a8e03651614b3c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 08:37:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ae1e0f51b4b4846a4fcf6392e37fb6c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d111f433752248bbb01dfb78148b3eff : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78ddca21131787c24d29fd442834a2c7c345fd7b","revisions":{"78ddca21131787c24d29fd442834a2c7c345fd7b":{"kind":"REWORK","_number":1,"created":"2026-09-14 05:09:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/1005498/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/98/1005498/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/98/1005498/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/98/1005498/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/98/1005498/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/98/1005498/1"}}},"commit":{"parents":[{"commit":"c3391a5710fe57c76d012a49d2beb0aba388bb47","subject":"Remove duplicated optional dependencies from test requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/c3391a5710fe57c76d012a49d2beb0aba388bb47"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-14 05:09:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-14 05:09:36.000000000","tz":540},"subject":"Ensure recursive detection of package files","message":"Ensure recursive detection of package files\n\nChange-Id: I96ed43eee662408ed6b26800afae74de3963c7e1\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/78ddca21131787c24d29fd442834a2c7c345fd7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/78ddca21131787c24d29fd442834a2c7c345fd7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
