)]}'
{"id":"openstack%2Foslo.log~528036","triplet_id":"openstack%2Foslo.log~master~I2e4e356d23dff2e6ac1946fe9e994c4d962ab55a","project":"openstack/oslo.log","branch":"master","topic":"bug/1729475","hashtags":[],"change_id":"I2e4e356d23dff2e6ac1946fe9e994c4d962ab55a","subject":"Truncate error_summary if exc_info not explicitly passed","status":"MERGED","created":"2017-12-14 18:42:31.000000000","updated":"2018-01-04 03:54:34.000000000","submitted":"2018-01-04 03:54:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"528036-1515038074503-75eacdef","meta_rev_id":"25b1e22a6c106712ab47ff7eeac85b34023f4cd0","_number":528036,"virtual_id_number":528036,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2018-01-04 03:54:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":2,"date":"2018-01-04 02:17:30.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":2,"date":"2018-01-03 21:22:50.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":1,"date":"2018-01-04 02:17:30.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-01-02 03:03:47.000000000","updated_by":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"reviewer":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"state":"REVIEWER"},{"updated":"2018-01-03 21:22:50.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2018-01-04 02:17:30.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2018-01-04 03:54:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"23fad0772b3f6e3652da6ed296b0c5457147a91b","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2017-12-14 18:42:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68b385d686dbee9334baeddd4faa49815099f6b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-14 20:54:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/36/528036/1/check/openstack-tox-pep8/612b8ae/ : SUCCESS in 3m 39s\n- openstack-tox-py27 http://logs.openstack.org/36/528036/1/check/openstack-tox-py27/c915a1f/ : SUCCESS in 6m 38s\n- openstack-tox-py35 http://logs.openstack.org/36/528036/1/check/openstack-tox-py35/d0e155b/ : SUCCESS in 3m 52s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/528036/1/check/build-openstack-sphinx-docs/c053a57/html/ : SUCCESS in 3m 41s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/36/528036/1/check/legacy-tempest-dsvm-neutron-src/a83dea2/ : SUCCESS in 2h 09m 09s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/36/528036/1/check/oslo.log-src-grenade-devstack/faf9f36/ : SUCCESS in 52m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a362996b16e4be3fb87ad1895912454662d7a438","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2017-12-25 03:18:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a3da44e420b58d47941cb0c827c8de93ce219fc7","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2018-01-02 03:04:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d6d74520c132e2d8699abac23e0287e82b7dfbeb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-02 04:28:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/36/528036/1/check/openstack-tox-pep8/0a56bb7/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/36/528036/1/check/openstack-tox-py27/bf78ada/ : SUCCESS in 5m 23s\n- openstack-tox-py35 http://logs.openstack.org/36/528036/1/check/openstack-tox-py35/50c7d49/ : SUCCESS in 5m 46s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/528036/1/check/build-openstack-sphinx-docs/49a2bbf/html/ : SUCCESS in 3m 21s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/36/528036/1/check/legacy-tempest-dsvm-neutron-src/65b0e9c/ : SUCCESS in 1h 17m 45s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/36/528036/1/check/oslo.log-src-grenade-devstack/0d3c1c9/ : SUCCESS in 47m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0aa1d309c6e3cc6124f82ba7fdd35959aa1715f0","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-01-02 14:29:20.000000000","message":"Patch Set 1: Code-Review-1\n\nOperators wanted exception details for all warning messages. If the problem is that the full traceback is being logged then we should fix that without removing the summary information.","accounts_in_message":[],"_revision_number":1},{"id":"6c992e153d9f32927b77696cbf1954b3ef5ea118","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2018-01-03 17:10:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e8e6cf3a97949dc8855e070a9d1254d8e7a7755b","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2018-01-03 17:56:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"177194f8bf6c21982ffdb6e6a28c924128201a43","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2018-01-03 17:56:39.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"0e9ca50c4508e1003765a31a032a309372f79ad7","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-01-03 18:03:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5a36b21440cbfd14c8f789b076365cd2d3b47c7a","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2018-01-03 20:39:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5a0558b98c7a513f1bab1839482ce28473cef9ef","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2018-01-03 21:14:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4110b601e5b78ed1f7f9407518b5aaf46cb7edc4","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-01-03 21:22:50.000000000","message":"Patch Set 2: Code-Review+2\n\nNice, this looks like it should solve the problem.","accounts_in_message":[],"_revision_number":2},{"id":"c107aeb7887a894e75183f2390dc86589f384194","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-03 22:37:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/36/528036/2/check/openstack-tox-pep8/b47391a/ : SUCCESS in 4m 44s\n- openstack-tox-py27 http://logs.openstack.org/36/528036/2/check/openstack-tox-py27/1415551/ : SUCCESS in 4m 54s\n- openstack-tox-py35 http://logs.openstack.org/36/528036/2/check/openstack-tox-py35/435ff17/ : SUCCESS in 5m 04s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/528036/2/check/build-openstack-sphinx-docs/71db282/html/ : SUCCESS in 3m 26s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/36/528036/2/check/legacy-tempest-dsvm-neutron-src/591ea77/ : SUCCESS in 1h 17m 05s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/36/528036/2/check/oslo.log-src-grenade-devstack/b5b0d72/ : SUCCESS in 46m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ba6e233d3ef97d8866e706d4f17db7577437695e","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2018-01-04 02:17:30.000000000","message":"Patch Set 2: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5ff9dd203f3708644232f34f46da2965121f2fa0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-04 02:17:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c52463cc3bb869c5f32a1fb966505a993198fd51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-04 03:54:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a491497f3c1d394f968e9a9e2f6b5e9a6e658c4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-04 03:54:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/36/528036/2/gate/openstack-tox-pep8/3680e81/ : SUCCESS in 3m 17s\n- openstack-tox-py27 http://logs.openstack.org/36/528036/2/gate/openstack-tox-py27/b646101/ : SUCCESS in 3m 48s\n- openstack-tox-py35 http://logs.openstack.org/36/528036/2/gate/openstack-tox-py35/65ee137/ : SUCCESS in 3m 28s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/528036/2/gate/build-openstack-sphinx-docs/1d7638f/html/ : SUCCESS in 3m 15s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/36/528036/2/gate/legacy-tempest-dsvm-neutron-src/5219a58/ : SUCCESS in 1h 35m 30s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a63438cb5b12e2a6973fb0e1b0b9ebce2f8116b6","revisions":{"ca53617b4f7f4f4de575bf5f617c8f62f3cf3e17":{"kind":"REWORK","_number":1,"created":"2017-12-14 18:42:31.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/36/528036/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/36/528036/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/36/528036/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/36/528036/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/36/528036/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/36/528036/1"}}},"commit":{"parents":[{"commit":"97434d48b296617a17af89601530c483c944e4b4","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/97434d48b296617a17af89601530c483c944e4b4"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2017-12-14 18:37:12.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2017-12-14 18:38:14.000000000","tz":0},"subject":"Omit error_summary if exc_info\u003dFalse was explicitly passed","message":"Omit error_summary if exc_info\u003dFalse was explicitly passed\n\nCurrently, when logging at level \u003e\u003d WARNING, exception information is\nadded to the log message even when exc_info\u003dFalse is explicitly passed.\nIn standard logging [1], if the exc_info keyword argument does not\nevaluate as false, it causes exception information to be added to the\nlogging message.\n\nThis adds a check for exc_info\u003dFalse and returns an empty string for\nthe error_summary in that case.\n\nCloses-Bug: #1729475\n\n[1] https://docs.python.org/3/library/logging.html#logging.Logger.debug\n\nChange-Id: I2e4e356d23dff2e6ac1946fe9e994c4d962ab55a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/ca53617b4f7f4f4de575bf5f617c8f62f3cf3e17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/ca53617b4f7f4f4de575bf5f617c8f62f3cf3e17"}]},"branch":"refs/heads/master"},"a63438cb5b12e2a6973fb0e1b0b9ebce2f8116b6":{"kind":"REWORK","_number":2,"created":"2018-01-03 21:14:49.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/36/528036/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/36/528036/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/36/528036/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/36/528036/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/36/528036/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/36/528036/2"}}},"commit":{"parents":[{"commit":"97434d48b296617a17af89601530c483c944e4b4","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/97434d48b296617a17af89601530c483c944e4b4"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2017-12-14 18:37:12.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2018-01-03 21:13:11.000000000","tz":0},"subject":"Truncate error_summary if exc_info not explicitly passed","message":"Truncate error_summary if exc_info not explicitly passed\n\nCurrently, when logging at level \u003e\u003d WARNING, exception information is\nadded to the log message as an \"error summary\" for operators. The error\nsummary includes the exception type and the message. However, when the\nexception is a _Remote exception generated from oslo.messaging and sent\nover RPC, the full traceback from the exception is appended to the\nexception message [1] to preserve it. So, the error summary ends up\noutputting the full traceback, unintentionally.\n\nThis truncates the error_summary to keep only the first line unless\nexc_info was explicitly passed, via LOG.exception(), for example.\n\nCloses-Bug: #1729475\n\n[1] https://github.com/openstack/oslo.messaging/blob/8d42484/oslo_messaging/_drivers/common.py#L208\n\nChange-Id: I2e4e356d23dff2e6ac1946fe9e994c4d962ab55a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a63438cb5b12e2a6973fb0e1b0b9ebce2f8116b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a63438cb5b12e2a6973fb0e1b0b9ebce2f8116b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
