)]}'
{"id":"openstack%2Foslo.log~610116","triplet_id":"openstack%2Foslo.log~master~Ida0ac3d88aa8bf04f9df275881e050d740508f58","project":"openstack/oslo.log","branch":"master","topic":"python3-train","hashtags":[],"change_id":"Ida0ac3d88aa8bf04f9df275881e050d740508f58","subject":"Add Python 3 Train unit tests","status":"ABANDONED","created":"2018-10-12 17:02:25.000000000","updated":"2019-07-16 20:40:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c8b949f7dfa6391b030699c005dacd75f31c5791","_number":610116,"virtual_id_number":610116,"owner":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":-1,"date":"2019-07-05 20:44:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":1,"date":"2019-07-08 14:37:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-10-15 14:47:38.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-06-20 21:30:18.000000000","updated_by":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"reviewer":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"state":"REVIEWER"},{"updated":"2019-07-05 20:44:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-08 14:37:28.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"}],"messages":[{"id":"c64d41950d4025b0f96a8ddc64443b8ee27fe4fb","author":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"date":"2018-10-12 17:02:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86ff97fba0d9c26b0eea9cfdd5035bbb0de52b94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-12 19:36:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/16/610116/1/check/openstack-tox-pep8/fcf225e/ : SUCCESS in 5m 49s\n- openstack-tox-py27 http://logs.openstack.org/16/610116/1/check/openstack-tox-py27/28a7902/ : SUCCESS in 6m 01s\n- openstack-tox-py37 http://logs.openstack.org/16/610116/1/check/openstack-tox-py37/2b51d16/ : FAILURE in 6m 33s\n- openstack-tox-docs http://logs.openstack.org/16/610116/1/check/openstack-tox-docs/ec557eb/html/ : SUCCESS in 4m 52s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/16/610116/1/check/legacy-tempest-dsvm-neutron-src/3005fcc/ : SUCCESS in 1h 40m 54s\n- tempest-full-py3 http://logs.openstack.org/16/610116/1/check/tempest-full-py3/2e7fde4/ : SUCCESS in 1h 30m 06s\n- build-openstack-releasenotes http://logs.openstack.org/16/610116/1/check/build-openstack-releasenotes/4d079da/html/ : SUCCESS in 5m 49s\n- openstack-tox-py36 http://logs.openstack.org/16/610116/1/check/openstack-tox-py36/24dd04e/ : SUCCESS in 5m 44s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/16/610116/1/check/oslo.log-src-grenade-devstack/105e88a/ : SUCCESS in 1h 04m 17s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/16/610116/1/check/oslo.log-jsonformatter/cb8b514/ : SUCCESS in 34m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/610116/1/check/openstack-tox-lower-constraints/1b08e7b/ : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"742ed6e9a670fa5cb2bdda9cdbe90054dca8a902","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2018-10-15 14:33:00.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"21d514eee08b27cc19c783220c7a4306a2617cde","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2018-10-15 14:47:38.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"273865b458a0d0395745ae306f9f92cc5f03999e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-15 19:28:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/16/610116/2/check/openstack-tox-pep8/0c10113/ : SUCCESS in 4m 37s\n- openstack-tox-py27 http://logs.openstack.org/16/610116/2/check/openstack-tox-py27/4118691/ : SUCCESS in 6m 02s\n- openstack-tox-py37 http://logs.openstack.org/16/610116/2/check/openstack-tox-py37/008ecda/ : FAILURE in 5m 37s\n- openstack-tox-docs http://logs.openstack.org/16/610116/2/check/openstack-tox-docs/eea8935/html/ : SUCCESS in 5m 45s\n- tempest-full http://logs.openstack.org/16/610116/2/check/tempest-full/773ca85/ : SUCCESS in 1h 30m 56s\n- tempest-full-py3 http://logs.openstack.org/16/610116/2/check/tempest-full-py3/68347c5/ : SUCCESS in 1h 15m 47s\n- build-openstack-releasenotes http://logs.openstack.org/16/610116/2/check/build-openstack-releasenotes/74fb71b/html/ : SUCCESS in 5m 33s\n- openstack-tox-py36 http://logs.openstack.org/16/610116/2/check/openstack-tox-py36/4f7d5c7/ : SUCCESS in 5m 08s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/16/610116/2/check/oslo.log-src-grenade-devstack/8670acb/ : SUCCESS in 48m 16s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/16/610116/2/check/oslo.log-jsonformatter/060590a/ : SUCCESS in 45m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/610116/2/check/openstack-tox-lower-constraints/c06f787/ : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":2},{"id":"63bbfc98274274bc9f27ecb26227fcdf3368222e","author":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"date":"2018-10-15 19:45:32.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"b58f9125f7b9eb5214a6ecbe1a1e1ecab23db482","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-06-20 21:30:18.000000000","message":"Patch Set 2:\n\nThis can be abandoned as this will be superceded by https://storyboard.openstack.org/#!/story/2005924","accounts_in_message":[],"_revision_number":2},{"id":"47b535d6ed0151ffc7c71b92847344d7bccf4c1a","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 18:46:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"17f7b90dcc7c85ffb042f8d8dc4b7e92c0cc57c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 20:44:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/16/610116/3/check/test-release-openstack/87be1bd/ : SUCCESS in 1m 43s\n- tempest-full http://logs.openstack.org/16/610116/3/check/tempest-full/49cbc3a/ : SUCCESS in 1h 50m 11s\n- tempest-full-py3 http://logs.openstack.org/16/610116/3/check/tempest-full-py3/18df293/ : SUCCESS in 1h 50m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/610116/3/check/openstack-tox-lower-constraints/bb09d3f/ : SUCCESS in 2m 51s\n- openstack-tox-pep8 http://logs.openstack.org/16/610116/3/check/openstack-tox-pep8/ccac0e4/ : SUCCESS in 2m 49s\n- openstack-tox-py27 http://logs.openstack.org/16/610116/3/check/openstack-tox-py27/2564267/ : SUCCESS in 2m 58s\n- openstack-tox-py36 http://logs.openstack.org/16/610116/3/check/openstack-tox-py36/079ce67/ : SUCCESS in 2m 42s\n- openstack-tox-py37 http://logs.openstack.org/16/610116/3/check/openstack-tox-py37/cb57a90/ : FAILURE in 3m 17s\n- openstack-tox-docs http://logs.openstack.org/16/610116/3/check/openstack-tox-docs/639300d/html/ : SUCCESS in 3m 02s\n- build-openstack-releasenotes http://logs.openstack.org/16/610116/3/check/build-openstack-releasenotes/59fd3c0/html/ : SUCCESS in 3m 32s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/16/610116/3/check/oslo.log-src-grenade-devstack/99669b5/ : SUCCESS in 57m 53s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/16/610116/3/check/oslo.log-jsonformatter/0777fbb/ : SUCCESS in 48m 29s","accounts_in_message":[],"_revision_number":3},{"id":"24b63e1a3aecfc4da5228eff52b1911269df00a5","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-07-08 14:37:28.000000000","message":"Patch Set 3: Code-Review+1\n\nPatch looks good but py37 test failure looks genuine.","accounts_in_message":[],"_revision_number":3},{"id":"c6cf931991107cb823efb0f6751b7087bca65d15","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-07-16 20:40:29.000000000","message":"Abandoned\n\nDuplicate of Iae43c960b980ba3c1aa83be8844ae2f873f9c186","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e1162969be0bf0e6922d9da43b910d565228d248","revisions":{"a448aa4c85fa1f327d816c2563bf4e79003cf932":{"kind":"REWORK","_number":1,"created":"2018-10-12 17:02:25.000000000","uploader":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"ref":"refs/changes/16/610116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/16/610116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/16/610116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/16/610116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/16/610116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/16/610116/1"}}},"commit":{"parents":[{"commit":"5cd02483c6839e987b723ecba389040a6d6fe79f","subject":"Always build universal wheels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/5cd02483c6839e987b723ecba389040a6d6fe79f"}]}],"author":{"name":"Charles Short","email":"chucks@redhat.com","date":"2018-10-12 17:00:09.000000000","tz":0},"committer":{"name":"Charles Short","email":"chucks@redhat.com","date":"2018-10-12 17:02:24.000000000","tz":0},"subject":"Change python3.5 job to python3.7 job on Stein+","message":"Change python3.5 job to python3.7 job on Stein+\n\npython3.5 was the only supported python3 version on Xenial, now that we have\nBionic Beaver nodes that support python3.7, lets switch to testing with\npython3.7 in addition with python3.6 in Stein and beyond.\nSee ML discussion here [1] for context.\n\n[1] http://lists.openstack.org/pipermail/openstack-dev/2018-October/135632.html\n\nChange-Id: Ida0ac3d88aa8bf04f9df275881e050d740508f58\nSigned-off-by: Charles Short \u003cchucks@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a448aa4c85fa1f327d816c2563bf4e79003cf932"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a448aa4c85fa1f327d816c2563bf4e79003cf932"}]},"branch":"refs/heads/master"},"65a23785c9d43358c933d8f1d4591ca8a3d17251":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2018-10-15 14:33:00.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/16/610116/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/16/610116/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/16/610116/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/16/610116/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/16/610116/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/16/610116/2"}}},"commit":{"parents":[{"commit":"5cd02483c6839e987b723ecba389040a6d6fe79f","subject":"Always build universal wheels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/5cd02483c6839e987b723ecba389040a6d6fe79f"}]}],"author":{"name":"Charles Short","email":"chucks@redhat.com","date":"2018-10-12 17:00:09.000000000","tz":0},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2018-10-15 14:32:57.000000000","tz":0},"subject":"Change python3.5 job to python3.7 job on Stein+","message":"Change python3.5 job to python3.7 job on Stein+\n\npython3.5 was the only supported python3 version on Xenial, now that we have\nBionic Beaver nodes that support python3.7, lets switch to testing with\npython3.7 in addition with python3.6 in Stein and beyond.\nSee ML discussion here [1] for context.\n\n[1] http://lists.openstack.org/pipermail/openstack-dev/2018-October/135632.html\n\nChange-Id: Ida0ac3d88aa8bf04f9df275881e050d740508f58\nSigned-off-by: Charles Short \u003cchucks@redhat.com\u003e\nStory: #2004073\nTask: #27440\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/65a23785c9d43358c933d8f1d4591ca8a3d17251"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/65a23785c9d43358c933d8f1d4591ca8a3d17251"}]},"branch":"refs/heads/master"},"e1162969be0bf0e6922d9da43b910d565228d248":{"kind":"REWORK","_number":3,"created":"2019-07-05 18:46:32.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/16/610116/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/16/610116/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/16/610116/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/16/610116/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/16/610116/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/16/610116/3"}}},"commit":{"parents":[{"commit":"e7a680e4d0542d21cc4272501f95572f8cd6c01e","subject":"Remove incubator migration docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/e7a680e4d0542d21cc4272501f95572f8cd6c01e"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 15:56:50.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 18:46:31.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: Ida0ac3d88aa8bf04f9df275881e050d740508f58\nStory: #2005924\nTask: #34234\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/e1162969be0bf0e6922d9da43b910d565228d248"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/e1162969be0bf0e6922d9da43b910d565228d248"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
