)]}'
{"id":"openstack%2Foslo.log~648393","triplet_id":"openstack%2Foslo.log~master~I1d7713ff7de930e40fc5b8ceb2cf2af54082c9cc","project":"openstack/oslo.log","branch":"master","topic":"bug/1822101","hashtags":[],"change_id":"I1d7713ff7de930e40fc5b8ceb2cf2af54082c9cc","subject":"Added cmdline information into fluentFormatter event message","status":"MERGED","created":"2019-03-28 13:02:45.000000000","updated":"2019-04-22 15:02:11.000000000","submitted":"2019-04-22 15:02:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"648393-1555945331686-d1a6a7d1","meta_rev_id":"0b5bcc6c25b6fafff0566d0909d2240237dcb7cc","_number":648393,"virtual_id_number":648393,"owner":{"_account_id":26768,"name":"Jan Vondra","email":"jan.vondra@ultimum.io","username":"JanVondra"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":2,"date":"2019-04-22 15:02:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-04-18 16:11:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-04-22 13:11:27.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":1,"date":"2019-04-16 10:36:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":1,"date":"2019-04-02 09:11:53.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2019-04-02 09:50:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-04-22 13:11:27.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-02 09:11:53.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-04-02 09:50:01.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2019-04-16 10:36:27.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-04-18 16:11:26.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-04-22 13:11:27.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2019-04-22 15:02:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f7e9a89ef86ced1420d5f53c5f7ca9d93de2abde","author":{"_account_id":26768,"name":"Jan Vondra","email":"jan.vondra@ultimum.io","username":"JanVondra"},"date":"2019-03-28 13:02:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"072ab54e4b4002ea7db0898a672408e898fd2c29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-28 14:37:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/93/648393/1/check/tempest-full/1e1efc6/ : SUCCESS in 1h 32m 55s\n- tempest-full-py3 http://logs.openstack.org/93/648393/1/check/tempest-full-py3/6e5ccb9/ : SUCCESS in 1h 21m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/93/648393/1/check/openstack-tox-lower-constraints/f5643d0/ : SUCCESS in 5m 08s\n- openstack-tox-pep8 http://logs.openstack.org/93/648393/1/check/openstack-tox-pep8/c20eecd/ : SUCCESS in 6m 48s\n- openstack-tox-py27 http://logs.openstack.org/93/648393/1/check/openstack-tox-py27/cedb7a2/ : SUCCESS in 5m 03s\n- openstack-tox-py35 http://logs.openstack.org/93/648393/1/check/openstack-tox-py35/0d19989/ : SUCCESS in 11m 51s\n- openstack-tox-py36 http://logs.openstack.org/93/648393/1/check/openstack-tox-py36/f0bbef0/ : SUCCESS in 5m 06s\n- openstack-tox-docs http://logs.openstack.org/93/648393/1/check/openstack-tox-docs/79a60e1/html/ : SUCCESS in 7m 16s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/93/648393/1/check/oslo.log-src-grenade-devstack/f565d6b/ : SUCCESS in 54m 42s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/93/648393/1/check/oslo.log-jsonformatter/8e2a308/ : SUCCESS in 38m 36s","accounts_in_message":[],"_revision_number":1},{"id":"43f5daa3707b8de12b82810092c718bd731bd614","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-04-02 09:11:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"21c8a6f878e21cdc00e6b61a11d54b9a8d994aa8","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2019-04-02 09:50:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"538bfe3eef6ba2b4e6d404ac6f420324702ce9b0","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-04-16 10:36:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"98b2f2a0366e64ce096194c5997fbbcf4e8eb90e","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-04-18 16:11:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4dfeb50ee2803bdf8f1563aa3c71334d503d83f8","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-04-22 13:11:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"604d9d03772f4ffeef92214390ca58c0a9843878","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 13:11:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f883b1950c28044391b6b27885feff4954ef1287","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 15:02:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"18d793199606dfb55ccd8c6757c072ea2df3bb95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 15:02:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/93/648393/1/gate/tempest-full/195d7bd/ : SUCCESS in 1h 40m 17s\n- tempest-full-py3 http://logs.openstack.org/93/648393/1/gate/tempest-full-py3/8456500/ : SUCCESS in 1h 45m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/93/648393/1/gate/openstack-tox-lower-constraints/2dd18a1/ : SUCCESS in 4m 56s\n- openstack-tox-pep8 http://logs.openstack.org/93/648393/1/gate/openstack-tox-pep8/a349821/ : SUCCESS in 4m 36s\n- openstack-tox-py27 http://logs.openstack.org/93/648393/1/gate/openstack-tox-py27/0ecbd4f/ : SUCCESS in 6m 10s\n- openstack-tox-py35 http://logs.openstack.org/93/648393/1/gate/openstack-tox-py35/0f235a3/ : SUCCESS in 4m 54s\n- openstack-tox-py36 http://logs.openstack.org/93/648393/1/gate/openstack-tox-py36/c60e2ef/ : SUCCESS in 5m 05s\n- openstack-tox-docs http://logs.openstack.org/93/648393/1/gate/openstack-tox-docs/79ded2f/html/ : SUCCESS in 4m 28s\n- oslo.log-jsonformatter http://logs.openstack.org/93/648393/1/gate/oslo.log-jsonformatter/ca98271/ : SUCCESS in 45m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ecbdb45c0a904c74f32bb757ddb99e915627fa1f","revisions":{"ecbdb45c0a904c74f32bb757ddb99e915627fa1f":{"kind":"REWORK","_number":1,"created":"2019-03-28 13:02:45.000000000","uploader":{"_account_id":26768,"name":"Jan Vondra","email":"jan.vondra@ultimum.io","username":"JanVondra"},"ref":"refs/changes/93/648393/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/93/648393/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/93/648393/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/93/648393/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/93/648393/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/93/648393/1"}}},"commit":{"parents":[{"commit":"ee2c228dd2b47c4946bb0bfa375336fd06ae575e","subject":"Replace openstack.org git:// URLs with https://","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/ee2c228dd2b47c4946bb0bfa375336fd06ae575e"}]}],"author":{"name":"Jan Vondra","email":"jan.vondra@ultimum.io","date":"2019-03-28 13:02:06.000000000","tz":60},"committer":{"name":"Jan Vondra","email":"jan.vondra@ultimum.io","date":"2019-03-28 13:02:06.000000000","tz":60},"subject":"Added cmdline information into fluentFormatter event message","message":"Added cmdline information into fluentFormatter event message\n\nSince process_name value defaults to \"MainProcess\" a new field cmdline\nhas been added in order to get information which process (daemon)\ntriggers the event.\n\nChange-Id: I1d7713ff7de930e40fc5b8ceb2cf2af54082c9cc\nCloses-Bug: 1822101\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/ecbdb45c0a904c74f32bb757ddb99e915627fa1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/ecbdb45c0a904c74f32bb757ddb99e915627fa1f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
