)]}'
{"id":"openstack%2Foslo.log~651378","triplet_id":"openstack%2Foslo.log~master~Ia5637f35315127c5cc46e01f5e38eccb094e772d","project":"openstack/oslo.log","branch":"master","topic":"deprecations","hashtags":[],"change_id":"Ia5637f35315127c5cc46e01f5e38eccb094e772d","subject":"Use raw string for regex","status":"MERGED","created":"2019-04-09 20:55:27.000000000","updated":"2019-05-04 01:27:39.000000000","submitted":"2019-05-04 01:27:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"651378-1556933259471-172fa3e6","meta_rev_id":"6844e3ff0eacb89e60ab276c45921e555c070531","_number":651378,"virtual_id_number":651378,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2019-05-04 01:27:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2019-05-03 16:33:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-04-16 10:34:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2019-05-03 16:33:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-16 10:34:18.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-05-03 16:33:54.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-04 01:27:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6c64821aa9323d38f301da8e547932799a2d4372","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-09 20:55:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b1c2b1d069dfadee03254ff6ee1c4aea4f94041","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-09 22:59:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/78/651378/1/check/tempest-full/4a406e6/ : SUCCESS in 1h 58m 35s\n- tempest-full-py3 http://logs.openstack.org/78/651378/1/check/tempest-full-py3/9cf54bf/ : SUCCESS in 1h 38m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/78/651378/1/check/openstack-tox-lower-constraints/c6451e3/ : SUCCESS in 5m 19s\n- openstack-tox-pep8 http://logs.openstack.org/78/651378/1/check/openstack-tox-pep8/d7d6490/ : SUCCESS in 4m 27s\n- openstack-tox-py27 http://logs.openstack.org/78/651378/1/check/openstack-tox-py27/3f1be5e/ : SUCCESS in 4m 54s\n- openstack-tox-py35 http://logs.openstack.org/78/651378/1/check/openstack-tox-py35/0f75045/ : SUCCESS in 4m 52s\n- openstack-tox-py36 http://logs.openstack.org/78/651378/1/check/openstack-tox-py36/fe5ec87/ : SUCCESS in 4m 55s\n- openstack-tox-docs http://logs.openstack.org/78/651378/1/check/openstack-tox-docs/b29ed12/html/ : SUCCESS in 4m 44s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/78/651378/1/check/oslo.log-src-grenade-devstack/caf75a3/ : SUCCESS in 1h 00m 13s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/78/651378/1/check/oslo.log-jsonformatter/d1ed122/ : SUCCESS in 36m 00s","accounts_in_message":[],"_revision_number":1},{"id":"3da48eb49506a9b20a4a9da43418f468cb3a0587","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-04-16 10:34:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5c0c34e5936753b46b7934cb7bd39ed98e50982b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 16:33:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nTrivial fix. Fast approving","accounts_in_message":[],"_revision_number":1},{"id":"952c14d5d6770e1ebe9f4b6d5a14b16f44becffd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 16:34:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8be5cf766640b78150fa15dc70b8d1339e2eccd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 18:42:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/78/651378/1/gate/tempest-full/caffd04/ : TIMED_OUT in 2h 06m 22s\n- tempest-full-py3 http://logs.openstack.org/78/651378/1/gate/tempest-full-py3/14bb624/ : SUCCESS in 1h 40m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/78/651378/1/gate/openstack-tox-lower-constraints/e516120/ : SUCCESS in 5m 16s\n- openstack-tox-pep8 http://logs.openstack.org/78/651378/1/gate/openstack-tox-pep8/7bee3e4/ : SUCCESS in 4m 11s\n- openstack-tox-py27 http://logs.openstack.org/78/651378/1/gate/openstack-tox-py27/ca47c08/ : SUCCESS in 5m 18s\n- openstack-tox-py36 http://logs.openstack.org/78/651378/1/gate/openstack-tox-py36/ad61d27/ : SUCCESS in 8m 05s\n- openstack-tox-docs http://logs.openstack.org/78/651378/1/gate/openstack-tox-docs/2e59f82/html/ : SUCCESS in 5m 10s\n- oslo.log-jsonformatter http://logs.openstack.org/78/651378/1/gate/oslo.log-jsonformatter/976923a/ : SUCCESS in 41m 51s","accounts_in_message":[],"_revision_number":1},{"id":"49bbf7351de1e8bc1f8e4d2be78aa66acb441a70","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 21:44:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fd9e094940e87087808a615a2f197a25083e1845","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 23:24:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/78/651378/1/check/tempest-full/c6c5a94/ : SUCCESS in 1h 37m 21s\n- tempest-full-py3 http://logs.openstack.org/78/651378/1/check/tempest-full-py3/b981cd7/ : SUCCESS in 1h 27m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/78/651378/1/check/openstack-tox-lower-constraints/89da5f2/ : SUCCESS in 4m 31s\n- openstack-tox-pep8 http://logs.openstack.org/78/651378/1/check/openstack-tox-pep8/4528edf/ : SUCCESS in 5m 40s\n- openstack-tox-py27 http://logs.openstack.org/78/651378/1/check/openstack-tox-py27/d8354ca/ : SUCCESS in 4m 26s\n- openstack-tox-py36 http://logs.openstack.org/78/651378/1/check/openstack-tox-py36/5c04c7f/ : SUCCESS in 4m 56s\n- openstack-tox-docs http://logs.openstack.org/78/651378/1/check/openstack-tox-docs/84bf2b8/html/ : SUCCESS in 5m 25s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/78/651378/1/check/oslo.log-src-grenade-devstack/6ecb848/ : SUCCESS in 1h 03m 28s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/78/651378/1/check/oslo.log-jsonformatter/0bc539e/ : SUCCESS in 43m 09s","accounts_in_message":[],"_revision_number":1},{"id":"c0f10034debc162c487b6cd683894720212f8abf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 23:24:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e56872dc82bf81911654ef4bcc94c341da960336","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 01:27:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/78/651378/1/gate/tempest-full/2acb782/ : SUCCESS in 1h 57m 51s\n- tempest-full-py3 http://logs.openstack.org/78/651378/1/gate/tempest-full-py3/d503066/ : SUCCESS in 1h 33m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/78/651378/1/gate/openstack-tox-lower-constraints/c3b446d/ : SUCCESS in 5m 02s\n- openstack-tox-pep8 http://logs.openstack.org/78/651378/1/gate/openstack-tox-pep8/493d3bd/ : SUCCESS in 4m 05s\n- openstack-tox-py27 http://logs.openstack.org/78/651378/1/gate/openstack-tox-py27/8a3d968/ : SUCCESS in 6m 26s\n- openstack-tox-py36 http://logs.openstack.org/78/651378/1/gate/openstack-tox-py36/a6e8017/ : SUCCESS in 5m 43s\n- openstack-tox-docs http://logs.openstack.org/78/651378/1/gate/openstack-tox-docs/732e948/html/ : SUCCESS in 6m 45s\n- oslo.log-jsonformatter http://logs.openstack.org/78/651378/1/gate/oslo.log-jsonformatter/a8868ee/ : SUCCESS in 43m 15s","accounts_in_message":[],"_revision_number":1},{"id":"fea765ebbf7fd063cd8cd2cb4da618a2393cbd7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 01:27:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b882c312465c311dfde059d6266efca2a43ec8e0","revisions":{"b882c312465c311dfde059d6266efca2a43ec8e0":{"kind":"REWORK","_number":1,"created":"2019-04-09 20:55:27.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/78/651378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/78/651378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/78/651378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/78/651378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/78/651378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/78/651378/1"}}},"commit":{"parents":[{"commit":"ee2c228dd2b47c4946bb0bfa375336fd06ae575e","subject":"Replace openstack.org git:// URLs with https://","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/ee2c228dd2b47c4946bb0bfa375336fd06ae575e"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-04-09 20:53:49.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-04-09 20:53:49.000000000","tz":-300},"subject":"Use raw string for regex","message":"Use raw string for regex\n\nSpecial characters in regular strings now cause DeprecationWarnings\nstarting with Python 3.6. This frequently happens with regex strings.\n\nOslo.log had one instance of a regex string containing \u0027\\\u0027 that would\ncause deprecation warnings in logs. This switches that instance over to\nbe a raw string to avoid these warnings.\n\nChange-Id: Ia5637f35315127c5cc46e01f5e38eccb094e772d\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/b882c312465c311dfde059d6266efca2a43ec8e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/b882c312465c311dfde059d6266efca2a43ec8e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
