)]}'
{"id":"openstack%2Foslo.log~652446","triplet_id":"openstack%2Foslo.log~master~Ib110f4a3c9475d2655dbf67d28474f54928b142a","project":"openstack/oslo.log","branch":"master","topic":"dropping-py35-testing","hashtags":[],"change_id":"Ib110f4a3c9475d2655dbf67d28474f54928b142a","subject":"Dropping the py35 testing","status":"MERGED","created":"2019-04-15 01:06:36.000000000","updated":"2019-05-03 18:30:14.000000000","submitted":"2019-05-03 18:30:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652446-1556908214058-76a6aaaf","meta_rev_id":"75805443a34e4445d1896425f524567e4111b3a7","_number":652446,"virtual_id_number":652446,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2019-05-03 18:30:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2019-05-03 16:33:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-04-16 10:33:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2019-05-03 16:33:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-16 10:33:54.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-05-03 16:33:36.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-03 18:30:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"97ecd8095f021d1d5b62698f5d88cc54983dae6f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-15 01:06:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0af86d472381bd7433c87c5dc10638fab0651f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 03:05:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/46/652446/1/check/test-release-openstack/bc84922/ : SUCCESS in 2m 53s\n- tempest-full http://logs.openstack.org/46/652446/1/check/tempest-full/44f0b49/ : SUCCESS in 1h 52m 50s\n- tempest-full-py3 http://logs.openstack.org/46/652446/1/check/tempest-full-py3/7620215/ : SUCCESS in 1h 37m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/46/652446/1/check/openstack-tox-lower-constraints/a84d931/ : SUCCESS in 4m 19s\n- openstack-tox-pep8 http://logs.openstack.org/46/652446/1/check/openstack-tox-pep8/b4ba5d3/ : SUCCESS in 4m 16s\n- openstack-tox-py27 http://logs.openstack.org/46/652446/1/check/openstack-tox-py27/2c5636c/ : SUCCESS in 4m 38s\n- openstack-tox-py36 http://logs.openstack.org/46/652446/1/check/openstack-tox-py36/7c9a4ad/ : SUCCESS in 5m 01s\n- openstack-tox-docs http://logs.openstack.org/46/652446/1/check/openstack-tox-docs/486a424/html/ : SUCCESS in 4m 29s\n- build-openstack-releasenotes http://logs.openstack.org/46/652446/1/check/build-openstack-releasenotes/877c964/html/ : SUCCESS in 3m 58s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/46/652446/1/check/oslo.log-src-grenade-devstack/7eb664e/ : SUCCESS in 1h 02m 33s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/46/652446/1/check/oslo.log-jsonformatter/06e2eae/ : SUCCESS in 39m 18s","accounts_in_message":[],"_revision_number":1},{"id":"018df9535f9f197ca60a06a0df8a101f33057a1d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-15 18:28:10.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2964188c7b678cfd7b9b87784cd4b6cdcd1b4959","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 20:21:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/46/652446/2/check/test-release-openstack/164c95e/ : SUCCESS in 4m 31s\n- tempest-full http://logs.openstack.org/46/652446/2/check/tempest-full/8568e9e/ : SUCCESS in 1h 42m 54s\n- tempest-full-py3 http://logs.openstack.org/46/652446/2/check/tempest-full-py3/d9de1d3/ : SUCCESS in 1h 48m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/46/652446/2/check/openstack-tox-lower-constraints/9832829/ : SUCCESS in 5m 49s\n- openstack-tox-pep8 http://logs.openstack.org/46/652446/2/check/openstack-tox-pep8/b577192/ : SUCCESS in 4m 58s\n- openstack-tox-py27 http://logs.openstack.org/46/652446/2/check/openstack-tox-py27/8f787fe/ : SUCCESS in 10m 17s\n- openstack-tox-py36 http://logs.openstack.org/46/652446/2/check/openstack-tox-py36/348a034/ : SUCCESS in 6m 25s\n- openstack-tox-docs http://logs.openstack.org/46/652446/2/check/openstack-tox-docs/653c467/html/ : SUCCESS in 6m 06s\n- build-openstack-releasenotes http://logs.openstack.org/46/652446/2/check/build-openstack-releasenotes/b8c2dae/html/ : SUCCESS in 5m 31s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/46/652446/2/check/oslo.log-src-grenade-devstack/9a28994/ : SUCCESS in 54m 07s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/46/652446/2/check/oslo.log-jsonformatter/392370d/ : SUCCESS in 40m 39s","accounts_in_message":[],"_revision_number":2},{"id":"8c3838e84cc9d7ed6115a2d92d067ef42f4634a6","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-04-16 10:33:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8dc407ff9e8ac3ab401b0944c8e6b8d47a6a914e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 16:33:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7baeace6fa287fa75aedc4c5569d67102e7ea64d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 16:33:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"96832696aa52149ba9bb535f62e952f26fb040be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 18:30:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/46/652446/2/gate/test-release-openstack/78249ae/ : SUCCESS in 5m 05s\n- tempest-full http://logs.openstack.org/46/652446/2/gate/tempest-full/ed47a63/ : SUCCESS in 1h 54m 41s\n- tempest-full-py3 http://logs.openstack.org/46/652446/2/gate/tempest-full-py3/cc72121/ : SUCCESS in 1h 38m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/46/652446/2/gate/openstack-tox-lower-constraints/84402da/ : SUCCESS in 5m 06s\n- openstack-tox-pep8 http://logs.openstack.org/46/652446/2/gate/openstack-tox-pep8/0338405/ : SUCCESS in 4m 55s\n- openstack-tox-py27 http://logs.openstack.org/46/652446/2/gate/openstack-tox-py27/089198b/ : SUCCESS in 5m 13s\n- openstack-tox-py36 http://logs.openstack.org/46/652446/2/gate/openstack-tox-py36/401af41/ : SUCCESS in 6m 26s\n- openstack-tox-docs http://logs.openstack.org/46/652446/2/gate/openstack-tox-docs/cc40150/html/ : SUCCESS in 8m 01s\n- build-openstack-releasenotes http://logs.openstack.org/46/652446/2/gate/build-openstack-releasenotes/0972331/html/ : SUCCESS in 7m 19s\n- oslo.log-jsonformatter http://logs.openstack.org/46/652446/2/gate/oslo.log-jsonformatter/98b6e9d/ : SUCCESS in 42m 32s","accounts_in_message":[],"_revision_number":2},{"id":"63bcc93814cb7336837b501dd681f09a90df1415","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 18:30:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6742d1f1c13dfbc71b6a462f8cbe9344602247b2","revisions":{"d56647f60dfc2973615e0518d525c1764b95b09a":{"kind":"REWORK","_number":1,"created":"2019-04-15 01:06:36.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/46/652446/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/46/652446/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/46/652446/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/46/652446/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/46/652446/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/46/652446/1"}}},"commit":{"parents":[{"commit":"ee2c228dd2b47c4946bb0bfa375336fd06ae575e","subject":"Replace openstack.org git:// URLs with https://","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/ee2c228dd2b47c4946bb0bfa375336fd06ae575e"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-15 01:06:15.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-15 01:06:34.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: Ib110f4a3c9475d2655dbf67d28474f54928b142a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/d56647f60dfc2973615e0518d525c1764b95b09a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/d56647f60dfc2973615e0518d525c1764b95b09a"}]},"branch":"refs/heads/master"},"6742d1f1c13dfbc71b6a462f8cbe9344602247b2":{"kind":"REWORK","_number":2,"created":"2019-04-15 18:28:10.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/46/652446/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/46/652446/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/46/652446/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/46/652446/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/46/652446/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/46/652446/2"}}},"commit":{"parents":[{"commit":"ee2c228dd2b47c4946bb0bfa375336fd06ae575e","subject":"Replace openstack.org git:// URLs with https://","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/ee2c228dd2b47c4946bb0bfa375336fd06ae575e"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-15 01:06:15.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-15 18:28:07.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: Ib110f4a3c9475d2655dbf67d28474f54928b142a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/6742d1f1c13dfbc71b6a462f8cbe9344602247b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/6742d1f1c13dfbc71b6a462f8cbe9344602247b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
