)]}'
{"id":"openstack%2Foslo.log~654732","triplet_id":"openstack%2Foslo.log~master~I35110a02e8b16f9c1d3b44441409db990eb1a23f","project":"openstack/oslo.log","branch":"master","hashtags":[],"change_id":"I35110a02e8b16f9c1d3b44441409db990eb1a23f","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-04-23 05:25:36.000000000","updated":"2019-05-05 19:53:33.000000000","submitted":"2019-05-05 19:53:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654732-1557086013995-172fa3e6","meta_rev_id":"6d7e636c09cb499513c52e211a6b372e99315ed4","_number":654732,"virtual_id_number":654732,"owner":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-05-05 19:53:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2019-05-05 18:09:52.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":1,"date":"2019-05-03 03:32:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":1,"date":"2019-04-30 09:21:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":2,"date":"2019-05-03 16:36:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-05-05 06:59:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":1,"date":"2019-05-05 18:09:52.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-30 09:21:51.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-05-03 03:32:31.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-05-03 16:36:09.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-05 06:59:37.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-05-05 18:09:52.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2019-05-05 19:53:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"af4bea84d48b93002e460e499adebf7568790d1b","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-04-23 05:25:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25612d107b14906315afad1e776208c448c546a5","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-04-23 05:26:25.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"87b892236b31145848741d0f303302fc65b7fb77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 09:22:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/32/654732/2/check/test-release-openstack/1280e49/ : SUCCESS in 3m 45s\n- tempest-full http://logs.openstack.org/32/654732/2/check/tempest-full/0fb6a5c/ : SUCCESS in 2h 04m 40s\n- tempest-full-py3 http://logs.openstack.org/32/654732/2/check/tempest-full-py3/a42d03d/ : SUCCESS in 1h 36m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/654732/2/check/openstack-tox-lower-constraints/871a7bb/ : SUCCESS in 5m 28s\n- openstack-tox-pep8 http://logs.openstack.org/32/654732/2/check/openstack-tox-pep8/eac9d6a/ : SUCCESS in 5m 50s\n- openstack-tox-py27 http://logs.openstack.org/32/654732/2/check/openstack-tox-py27/06fa630/ : SUCCESS in 5m 32s\n- openstack-tox-py35 http://logs.openstack.org/32/654732/2/check/openstack-tox-py35/c1fa0ff/ : SUCCESS in 5m 32s\n- openstack-tox-py36 http://logs.openstack.org/32/654732/2/check/openstack-tox-py36/200f221/ : SUCCESS in 5m 37s\n- openstack-tox-docs http://logs.openstack.org/32/654732/2/check/openstack-tox-docs/eeb3bf5/html/ : SUCCESS in 4m 41s\n- build-openstack-releasenotes http://logs.openstack.org/32/654732/2/check/build-openstack-releasenotes/2e56d4c/html/ : SUCCESS in 5m 15s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/32/654732/2/check/oslo.log-src-grenade-devstack/0e61911/ : SUCCESS in 1h 09m 16s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/32/654732/2/check/oslo.log-jsonformatter/77a847c/ : SUCCESS in 48m 11s","accounts_in_message":[],"_revision_number":2},{"id":"4cce5d3877dbcc5de8a3ffe1fe4a26a1bdf30b83","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-04-26 02:05:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"17815b4423c46fa98edfd2845017a79ac99ae3f9","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-04-26 02:06:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b47d891db19a8c4a0ad5cdae0eebe6698b3e9d58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 04:01:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/32/654732/4/check/test-release-openstack/a920e3e/ : SUCCESS in 3m 06s\n- tempest-full http://logs.openstack.org/32/654732/4/check/tempest-full/f52e868/ : SUCCESS in 1h 47m 28s\n- tempest-full-py3 http://logs.openstack.org/32/654732/4/check/tempest-full-py3/e8c55c0/ : SUCCESS in 1h 26m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/654732/4/check/openstack-tox-lower-constraints/1486a28/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/32/654732/4/check/openstack-tox-pep8/fc83452/ : SUCCESS in 4m 36s\n- openstack-tox-py27 http://logs.openstack.org/32/654732/4/check/openstack-tox-py27/ff92301/ : SUCCESS in 3m 50s\n- openstack-tox-py35 http://logs.openstack.org/32/654732/4/check/openstack-tox-py35/5faa67a/ : SUCCESS in 8m 40s\n- openstack-tox-py36 http://logs.openstack.org/32/654732/4/check/openstack-tox-py36/4f92d95/ : SUCCESS in 4m 47s\n- openstack-tox-docs http://logs.openstack.org/32/654732/4/check/openstack-tox-docs/5735f1e/html/ : SUCCESS in 4m 04s\n- build-openstack-releasenotes http://logs.openstack.org/32/654732/4/check/build-openstack-releasenotes/53491df/html/ : SUCCESS in 3m 56s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/32/654732/4/check/oslo.log-src-grenade-devstack/6b5f4de/ : SUCCESS in 57m 45s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/32/654732/4/check/oslo.log-jsonformatter/8642284/ : SUCCESS in 39m 45s","accounts_in_message":[],"_revision_number":4},{"id":"88d1baeb2ea13855ee2b6567ba1c1d09096300cf","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-30 05:33:31.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"00e1f284cb95a68a28b91b400a2b287cb3550460","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-04-30 05:38:20.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7deee6854b99f1263cff93976e65d44d274c5875","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 07:35:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/32/654732/5/check/test-release-openstack/56b2952/ : SUCCESS in 3m 09s\n- tempest-full http://logs.openstack.org/32/654732/5/check/tempest-full/fc1e340/ : SUCCESS in 1h 51m 02s\n- tempest-full-py3 http://logs.openstack.org/32/654732/5/check/tempest-full-py3/d1f5d64/ : SUCCESS in 1h 53m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/654732/5/check/openstack-tox-lower-constraints/d793fb5/ : SUCCESS in 5m 18s\n- openstack-tox-pep8 http://logs.openstack.org/32/654732/5/check/openstack-tox-pep8/fef3dd7/ : SUCCESS in 5m 01s\n- openstack-tox-py27 http://logs.openstack.org/32/654732/5/check/openstack-tox-py27/c976b6b/ : SUCCESS in 5m 21s\n- openstack-tox-py35 http://logs.openstack.org/32/654732/5/check/openstack-tox-py35/9c863ea/ : SUCCESS in 5m 49s\n- openstack-tox-py36 http://logs.openstack.org/32/654732/5/check/openstack-tox-py36/c482918/ : SUCCESS in 5m 57s\n- openstack-tox-docs http://logs.openstack.org/32/654732/5/check/openstack-tox-docs/2c90738/html/ : SUCCESS in 4m 08s\n- build-openstack-releasenotes http://logs.openstack.org/32/654732/5/check/build-openstack-releasenotes/98cac1f/html/ : SUCCESS in 3m 51s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/32/654732/5/check/oslo.log-src-grenade-devstack/2186829/ : SUCCESS in 1h 03m 45s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/32/654732/5/check/oslo.log-jsonformatter/3fe588a/ : SUCCESS in 41m 53s","accounts_in_message":[],"_revision_number":5},{"id":"6e109356374cf564e655e179f41fdd9b15c523f2","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-30 09:21:51.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"c5ebdcdfda0fa15a095b803fe6f53991eb1c5356","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-05-03 03:32:31.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"e662facc60ad1a83281536bc2ffdcb04f281e4be","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 16:36:09.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"082bfe278ff992dcbced1ad8b03784cde8a84343","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-05 06:59:37.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"4bab2222bfaa5092c186575b6dde44eca4d309c5","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-05-05 18:09:52.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"95729117c1daa337d79b57fe098b17043f0b98ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 18:10:04.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"55a7b195483d04de57a35acd2a22a2a3c965ac82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 19:53:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"6d7e636c09cb499513c52e211a6b372e99315ed4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 19:53:33.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/32/654732/5/gate/test-release-openstack/e659e35/ : SUCCESS in 2m 50s\n- tempest-full http://logs.openstack.org/32/654732/5/gate/tempest-full/a468915/ : SUCCESS in 1h 31m 47s\n- tempest-full-py3 http://logs.openstack.org/32/654732/5/gate/tempest-full-py3/284abc7/ : SUCCESS in 1h 42m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/654732/5/gate/openstack-tox-lower-constraints/5e75f5e/ : SUCCESS in 4m 45s\n- openstack-tox-pep8 http://logs.openstack.org/32/654732/5/gate/openstack-tox-pep8/859d6cc/ : SUCCESS in 4m 04s\n- openstack-tox-py27 http://logs.openstack.org/32/654732/5/gate/openstack-tox-py27/c30609a/ : SUCCESS in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/32/654732/5/gate/openstack-tox-py36/edf5feb/ : SUCCESS in 4m 28s\n- openstack-tox-docs http://logs.openstack.org/32/654732/5/gate/openstack-tox-docs/41f7a2e/html/ : SUCCESS in 5m 45s\n- build-openstack-releasenotes http://logs.openstack.org/32/654732/5/gate/build-openstack-releasenotes/eeae1c9/html/ : SUCCESS in 4m 17s\n- oslo.log-jsonformatter http://logs.openstack.org/32/654732/5/gate/oslo.log-jsonformatter/bcaceb1/ : SUCCESS in 31m 13s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"815df26f0ddcc1a7a9fa3939d6f7c8b714314270","revisions":{"96579d3b3361496863d65404dab9a4d2c68910c5":{"kind":"REWORK","_number":1,"created":"2019-04-23 05:25:36.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/32/654732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/32/654732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/32/654732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/32/654732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/32/654732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/32/654732/1"}}},"commit":{"parents":[{"commit":"2d1aa7254b01976291c312aceb6ab691985b38ff","subject":"Merge \"Added cmdline information into fluentFormatter event message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/2d1aa7254b01976291c312aceb6ab691985b38ff"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-04-23 05:25:13.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-04-23 05:25:33.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I35110a02e8b16f9c1d3b44441409db990eb1a23f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/96579d3b3361496863d65404dab9a4d2c68910c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/96579d3b3361496863d65404dab9a4d2c68910c5"}]},"branch":"refs/heads/master"},"3f23d6f9126284f6545e22acff9ac9b5415e2389":{"kind":"REWORK","_number":2,"created":"2019-04-23 05:26:25.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/32/654732/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/32/654732/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/32/654732/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/32/654732/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/32/654732/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/32/654732/2"}}},"commit":{"parents":[{"commit":"2d1aa7254b01976291c312aceb6ab691985b38ff","subject":"Merge \"Added cmdline information into fluentFormatter event message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/2d1aa7254b01976291c312aceb6ab691985b38ff"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-04-23 05:25:13.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-04-23 05:26:14.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I35110a02e8b16f9c1d3b44441409db990eb1a23f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/3f23d6f9126284f6545e22acff9ac9b5415e2389"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/3f23d6f9126284f6545e22acff9ac9b5415e2389"}]},"branch":"refs/heads/master"},"80c346ef926ce79a14fa6f8a9b9d6db17584f84e":{"kind":"REWORK","_number":3,"created":"2019-04-26 02:05:47.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/32/654732/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/32/654732/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/32/654732/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/32/654732/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/32/654732/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/32/654732/3"}}},"commit":{"parents":[{"commit":"2d1aa7254b01976291c312aceb6ab691985b38ff","subject":"Merge \"Added cmdline information into fluentFormatter event message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/2d1aa7254b01976291c312aceb6ab691985b38ff"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-04-23 05:25:13.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-04-26 02:05:40.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I35110a02e8b16f9c1d3b44441409db990eb1a23f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/80c346ef926ce79a14fa6f8a9b9d6db17584f84e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/80c346ef926ce79a14fa6f8a9b9d6db17584f84e"}]},"branch":"refs/heads/master"},"5857ae957dbe5790886ace75f816a3202f208735":{"kind":"REWORK","_number":4,"created":"2019-04-26 02:06:57.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/32/654732/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/32/654732/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/32/654732/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/32/654732/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/32/654732/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/32/654732/4"}}},"commit":{"parents":[{"commit":"2d1aa7254b01976291c312aceb6ab691985b38ff","subject":"Merge \"Added cmdline information into fluentFormatter event message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/2d1aa7254b01976291c312aceb6ab691985b38ff"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-04-23 05:25:13.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-04-26 02:06:50.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I35110a02e8b16f9c1d3b44441409db990eb1a23f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/5857ae957dbe5790886ace75f816a3202f208735"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/5857ae957dbe5790886ace75f816a3202f208735"}]},"branch":"refs/heads/master"},"815df26f0ddcc1a7a9fa3939d6f7c8b714314270":{"kind":"REWORK","_number":5,"created":"2019-04-30 05:38:20.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/32/654732/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/32/654732/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/32/654732/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/32/654732/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/32/654732/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/32/654732/5"}}},"commit":{"parents":[{"commit":"2d1aa7254b01976291c312aceb6ab691985b38ff","subject":"Merge \"Added cmdline information into fluentFormatter event message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/2d1aa7254b01976291c312aceb6ab691985b38ff"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-04-23 05:25:13.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-04-30 05:38:11.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I35110a02e8b16f9c1d3b44441409db990eb1a23f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/815df26f0ddcc1a7a9fa3939d6f7c8b714314270"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/815df26f0ddcc1a7a9fa3939d6f7c8b714314270"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
