)]}'
{"id":"openstack%2Foslo.log~655645","triplet_id":"openstack%2Foslo.log~master~Ifcb555c8f4ccb813b48c613c79dceb4bf23318eb","project":"openstack/oslo.log","branch":"master","topic":"opendev-u-c","hashtags":[],"change_id":"Ifcb555c8f4ccb813b48c613c79dceb4bf23318eb","subject":"Use upper constraints from opendev.org","status":"ABANDONED","created":"2019-04-25 09:45:46.000000000","updated":"2019-08-01 15:19:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3567139a5449bcd3420fbf7803cf2f87d8d663e0","_number":655645,"virtual_id_number":655645,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":1,"date":"2019-04-25 11:33:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"disliked":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"all":[{"value":-1,"date":"2019-05-07 11:22:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":2,"date":"2019-05-03 16:38:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-05-03 00:00:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-25 11:33:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-03 00:00:23.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-05-03 16:38:29.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-07 11:22:18.000000000","updated_by":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"reviewer":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"state":"REVIEWER"}],"messages":[{"id":"ea7fa5ee43baba1cb218459e444d1c3b79933dc0","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2019-04-25 09:45:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"435499240ffcb960d19a4f1bc7d1dbfd707658b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 11:33:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/45/655645/1/check/tempest-full/c1fd852/ : SUCCESS in 1h 43m 29s\n- tempest-full-py3 http://logs.openstack.org/45/655645/1/check/tempest-full-py3/c1d94a1/ : SUCCESS in 1h 34m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/655645/1/check/openstack-tox-lower-constraints/5c2fdc7/ : SUCCESS in 5m 15s\n- openstack-tox-pep8 http://logs.openstack.org/45/655645/1/check/openstack-tox-pep8/ebaa49e/ : SUCCESS in 4m 37s\n- openstack-tox-py27 http://logs.openstack.org/45/655645/1/check/openstack-tox-py27/179c1dc/ : SUCCESS in 4m 25s\n- openstack-tox-py35 http://logs.openstack.org/45/655645/1/check/openstack-tox-py35/665d7fe/ : SUCCESS in 4m 29s\n- openstack-tox-py36 http://logs.openstack.org/45/655645/1/check/openstack-tox-py36/daf9536/ : SUCCESS in 4m 28s\n- openstack-tox-docs http://logs.openstack.org/45/655645/1/check/openstack-tox-docs/a0bf288/html/ : SUCCESS in 4m 26s\n- build-openstack-releasenotes http://logs.openstack.org/45/655645/1/check/build-openstack-releasenotes/b560e01/html/ : SUCCESS in 4m 23s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/45/655645/1/check/oslo.log-src-grenade-devstack/6d90a7c/ : SUCCESS in 49m 57s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/45/655645/1/check/oslo.log-jsonformatter/0896833/ : SUCCESS in 55m 12s","accounts_in_message":[],"_revision_number":1},{"id":"82755f1b19b2a014f19a2c21efb9132f1f2eacfb","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-05-03 00:00:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d02bc448217dfc4dfd26a826e4233007e2ba168a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 16:38:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"298bfbbe3bcaaa46d11c47721db90de085c26320","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-05-07 11:22:18.000000000","message":"Patch Set 1: Code-Review-1\n\nThat already in master branch.","accounts_in_message":[],"_revision_number":1},{"id":"45e31cc00a9795db6cc441a51d4a1d7c2f1fc54b","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-08-01 14:59:00.000000000","message":"Patch Set 1:\n\nThis patch can be abandoned.","accounts_in_message":[],"_revision_number":1},{"id":"230da5f6cbe89b4c3f15da95e3128c3c9c6640c0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-08-01 15:19:25.000000000","message":"Abandoned\n\nHello,\n\nFirst, thanks for these changes.\n\nSimilar fixes was already merged few months ago \u003d\u003e I73e284393ce26c91ea675d5be213ff5e910b4dfe\n\nhttps://github.com/openstack/oslo.log/commit/d693dc224788d6345c53352687308f097cf86618\n\nhttps://github.com/openstack/oslo.log/commit/815df26f0ddcc1a7a9fa3939d6f7c8b714314270\n\nI think we don\u0027t need to change again.\n\nI abandon this one for these reasons.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6eceb160eaaebbc6ea04768f62876738aa61c2d1","revisions":{"6eceb160eaaebbc6ea04768f62876738aa61c2d1":{"kind":"REWORK","_number":1,"created":"2019-04-25 09:45:46.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/45/655645/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/45/655645/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/45/655645/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/45/655645/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/45/655645/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/45/655645/1"}}},"commit":{"parents":[{"commit":"2d1aa7254b01976291c312aceb6ab691985b38ff","subject":"Merge \"Added cmdline information into fluentFormatter event message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/2d1aa7254b01976291c312aceb6ab691985b38ff"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2019-04-25 09:39:36.000000000","tz":180},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2019-04-25 09:45:40.000000000","tz":180},"subject":"Use upper constraints from opendev.org","message":"Use upper constraints from opendev.org\n\noslo.cache uses a simple curl command in its tox_install.sh\nto fetch the upper-constraints, and curl does not follow redirects\nthat are now in place due to move to opendev.org.\n\nThis breaks running tox-based tests w/o overriding the\nUPPER_CONSTRAINTS_FILE which upstream CI does,\nfor example running tox locally.\n\nChange-Id: Ifcb555c8f4ccb813b48c613c79dceb4bf23318eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/6eceb160eaaebbc6ea04768f62876738aa61c2d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/6eceb160eaaebbc6ea04768f62876738aa61c2d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
