)]}'
{"id":"openstack%2Foslo.log~655647","triplet_id":"openstack%2Foslo.log~stable%2Fstein~Ifcb555c8f4ccb813b48c613c79dceb4bf23318eb","project":"openstack/oslo.log","branch":"stable/stein","topic":"opendev-u-c","hashtags":[],"change_id":"Ifcb555c8f4ccb813b48c613c79dceb4bf23318eb","subject":"Use upper constraints from opendev.org","status":"ABANDONED","created":"2019-04-25 09:48:40.000000000","updated":"2019-12-17 19:02:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"499b14b079edb80b9d67ef690340958b52bbea1f","_number":655647,"virtual_id_number":655647,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"rejected":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":-2,"date":"2019-05-20 09:33:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"recommended":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"all":[{"value":-2,"date":"2019-05-22 18:22:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-05-20 07:42:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":2,"date":"2019-05-03 16:38:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-05-03 00:00:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-20 03:39:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-05-20 07:42:46.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-03 00:00:10.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-05-03 16:38:40.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-20 03:39:29.000000000","updated_by":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"reviewer":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"state":"REVIEWER"},{"updated":"2019-05-20 07:42:46.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2019-05-20 09:33:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-22 18:22:35.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"34d829ccf51065db4b982cfbf0a84b79255868c3","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2019-04-25 09:48:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb719bea98f4e14d43a812d4607d3cb84b655080","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 11:49:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/47/655647/1/check/tempest-full/c19ee7e/ : SUCCESS in 1h 48m 31s\n- tempest-full-py3 http://logs.openstack.org/47/655647/1/check/tempest-full-py3/cdb889b/ : SUCCESS in 1h 38m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/655647/1/check/openstack-tox-lower-constraints/ec2f35d/ : SUCCESS in 4m 52s\n- openstack-tox-pep8 http://logs.openstack.org/47/655647/1/check/openstack-tox-pep8/1cec3ce/ : SUCCESS in 4m 46s\n- openstack-tox-py27 http://logs.openstack.org/47/655647/1/check/openstack-tox-py27/0451ce2/ : SUCCESS in 4m 13s\n- openstack-tox-py35 http://logs.openstack.org/47/655647/1/check/openstack-tox-py35/2ead6a9/ : SUCCESS in 5m 04s\n- openstack-tox-py36 http://logs.openstack.org/47/655647/1/check/openstack-tox-py36/7fd3da1/ : SUCCESS in 4m 12s\n- openstack-tox-docs http://logs.openstack.org/47/655647/1/check/openstack-tox-docs/76acea3/html/ : SUCCESS in 4m 35s\n- build-openstack-releasenotes http://logs.openstack.org/47/655647/1/check/build-openstack-releasenotes/765fecd/html/ : SUCCESS in 4m 29s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/47/655647/1/check/oslo.log-src-grenade-devstack/8ce54cb/ : SUCCESS in 57m 53s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/47/655647/1/check/oslo.log-jsonformatter/372f058/ : SUCCESS in 41m 48s","accounts_in_message":[],"_revision_number":1},{"id":"365cea9b4823e70721076072bc10f68aa5bc3a14","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-05-03 00:00:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f636ac59c8195670ad2503721f3629e30351a270","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 16:38:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4ea7f2dd53a4d36c739226b84d354fef7bdcd6c9","author":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"date":"2019-05-20 03:39:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dd9840e336342ca31e44368a01f023df1ceded86","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2019-05-20 07:42:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a6d0ee1f95574e97a40c07496816f02aa544bcc1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 07:42:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"00709588f97df14ec1102d9408d557fc1306c124","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 09:33:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/47/655647/1/gate/tempest-full/7b12831/ : SUCCESS in 1h 49m 54s\n- tempest-full-py3 http://logs.openstack.org/47/655647/1/gate/tempest-full-py3/32976bc/ : SUCCESS in 1h 28m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/655647/1/gate/openstack-tox-lower-constraints/37d1efe/ : SUCCESS in 5m 47s\n- openstack-tox-pep8 http://logs.openstack.org/47/655647/1/gate/openstack-tox-pep8/77a616a/ : FAILURE in 4m 07s\n- openstack-tox-py27 http://logs.openstack.org/47/655647/1/gate/openstack-tox-py27/5d37265/ : SUCCESS in 4m 12s\n- openstack-tox-py35 http://logs.openstack.org/47/655647/1/gate/openstack-tox-py35/602ad77/ : SUCCESS in 4m 40s\n- openstack-tox-py36 http://logs.openstack.org/47/655647/1/gate/openstack-tox-py36/df9c077/ : SUCCESS in 5m 44s\n- openstack-tox-docs http://logs.openstack.org/47/655647/1/gate/openstack-tox-docs/02a575b/html/ : SUCCESS in 4m 02s\n- build-openstack-releasenotes http://logs.openstack.org/47/655647/1/gate/build-openstack-releasenotes/1327e46/html/ : SUCCESS in 3m 59s\n- oslo.log-jsonformatter http://logs.openstack.org/47/655647/1/gate/oslo.log-jsonformatter/6adfe45/ : SUCCESS in 37m 53s","accounts_in_message":[],"_revision_number":1},{"id":"095443a22542f8a1a05797e6738e0e01638eb681","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-22 18:22:35.000000000","message":"Patch Set 1: Code-Review-2\n\nThis is not correct, we do not want the u-c URL changed to this.","accounts_in_message":[],"_revision_number":1},{"id":"d805685e295c9af06c0d6528e8c2190003a0b0e4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-17 19:02:09.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"462695f37b92592c5cd1e1db89f7d8a9b3f13eab","revisions":{"462695f37b92592c5cd1e1db89f7d8a9b3f13eab":{"kind":"REWORK","_number":1,"created":"2019-04-25 09:48:40.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/47/655647/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/47/655647/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/47/655647/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/47/655647/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/47/655647/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/47/655647/1"}}},"commit":{"parents":[{"commit":"f266cf6ddeb98cf8432fc5fff692fe5fe6c3c6d2","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/f266cf6ddeb98cf8432fc5fff692fe5fe6c3c6d2"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2019-04-25 09:39:36.000000000","tz":180},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2019-04-25 09:48:09.000000000","tz":180},"subject":"Use upper constraints from opendev.org","message":"Use upper constraints from opendev.org\n\noslo.cache uses a simple curl command in its tox_install.sh\nto fetch the upper-constraints, and curl does not follow redirects\nthat are now in place due to move to opendev.org.\n\nThis breaks running tox-based tests w/o overriding the\nUPPER_CONSTRAINTS_FILE which upstream CI does,\nfor example running tox locally.\n\nChange-Id: Ifcb555c8f4ccb813b48c613c79dceb4bf23318eb\n(cherry picked from commit 6eceb160eaaebbc6ea04768f62876738aa61c2d1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/462695f37b92592c5cd1e1db89f7d8a9b3f13eab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/462695f37b92592c5cd1e1db89f7d8a9b3f13eab"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
