)]}'
{"id":"openstack%2Foslo.log~655648","triplet_id":"openstack%2Foslo.log~stable%2Frocky~Ifcb555c8f4ccb813b48c613c79dceb4bf23318eb","project":"openstack/oslo.log","branch":"stable/rocky","topic":"opendev-u-c","hashtags":[],"change_id":"Ifcb555c8f4ccb813b48c613c79dceb4bf23318eb","subject":"Use upper constraints from opendev.org","status":"MERGED","created":"2019-04-25 09:49:48.000000000","updated":"2019-12-17 21:54:37.000000000","submitted":"2019-12-17 21:52:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655648-1576619525091-0d4ca9dc","meta_rev_id":"ac5fee2fd351cc8f2ba2e40d986e7ea3a5c1343b","_number":655648,"virtual_id_number":655648,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-17 21:52:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":2,"date":"2019-05-20 07:47:27.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":2,"date":"2019-05-03 16:38:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-05-03 00:00:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":1,"date":"2019-05-20 07:47:27.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-03 00:00:19.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-05-03 16:38:35.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-20 07:47:27.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2019-12-17 21:52:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae9697f441f73a0cad85014ca8ed15fb28e28f55","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2019-04-25 09:49:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f635d76afbdb5b52a94526b90f95051db1f13063","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 11:46:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/48/655648/1/check/openstack-tox-pep8/87e6146/ : SUCCESS in 7m 07s\n- openstack-tox-py27 http://logs.openstack.org/48/655648/1/check/openstack-tox-py27/82ca4af/ : SUCCESS in 4m 45s\n- openstack-tox-py35 http://logs.openstack.org/48/655648/1/check/openstack-tox-py35/9c381f8/ : SUCCESS in 5m 02s\n- openstack-tox-docs http://logs.openstack.org/48/655648/1/check/openstack-tox-docs/a658109/html/ : SUCCESS in 4m 55s\n- tempest-full http://logs.openstack.org/48/655648/1/check/tempest-full/bc10c66/ : SUCCESS in 1h 33m 32s\n- build-openstack-releasenotes http://logs.openstack.org/48/655648/1/check/build-openstack-releasenotes/792925a/html/ : SUCCESS in 3m 53s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/48/655648/1/check/oslo.log-src-grenade-devstack/0e4b456/ : SUCCESS in 50m 19s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/48/655648/1/check/openstack-tox-lower-constraints/71a25a2/ : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":1},{"id":"219e7f6e77fd0879172211dd3e60459e463584ed","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-05-03 00:00:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"23f363d7f3a4ddc90407a78750a4cf51c45df511","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 16:38:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e62a4adc30f4697362592df33b92c26260b10895","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2019-05-20 07:47:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fd5e2cc4c45a0e431477b9ebea17ab513800dfcd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 07:47:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b1c548e0e84714f00fd3c7de050b21243783aa2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 09:32:51.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/48/655648/1/gate/openstack-tox-pep8/0a6001f/ : FAILURE in 4m 35s\n- openstack-tox-py27 http://logs.openstack.org/48/655648/1/gate/openstack-tox-py27/8c4ff60/ : SUCCESS in 6m 33s\n- openstack-tox-py35 http://logs.openstack.org/48/655648/1/gate/openstack-tox-py35/de3f994/ : SUCCESS in 5m 08s\n- openstack-tox-docs http://logs.openstack.org/48/655648/1/gate/openstack-tox-docs/0735972/html/ : SUCCESS in 6m 48s\n- tempest-full http://logs.openstack.org/48/655648/1/gate/tempest-full/9d2d6ab/ : FAILURE in 1h 24m 50s\n- build-openstack-releasenotes http://logs.openstack.org/48/655648/1/gate/build-openstack-releasenotes/ce87eb6/html/ : SUCCESS in 4m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/48/655648/1/gate/openstack-tox-lower-constraints/78ab447/ : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d7f781d93fb8864ac9b2e2b73738bb1ef3ed8488","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-17 19:01:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"47d80e940434c577279c17a62693820f97c7a85c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 20:28:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/392c1d56e7454067b6e31bbe9d46b7f8 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f82e314172db44b2ad8c62a3f892c11e : SUCCESS in 4m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a1a4c824abfe4fadb14be942ba8d941c : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c8aabf237a24488a2171a4d2e7168e7 : SUCCESS in 5m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/adea425790244d83b541c09729abb2c1 : SUCCESS in 1h 25m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d93941bc3d1c4e589703a83743193496 : SUCCESS in 4m 36s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/b6e3d38bf68541ef8b08d49a7c0ede9a : SUCCESS in 50m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43979c7bf2f64f6e85ddc6d090fe1790 : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":1},{"id":"20b64f3a15c5014c2c4e18a00b4ee05c0e574505","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 20:29:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"714b37effb92f48443985f93c2fe001c3efa1e35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 21:52:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81a298b43562492f9bc9f75d80930e49 : SUCCESS in 6m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa78305e52104b078cd5c8ef48b1df6f : SUCCESS in 4m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2aeb676fc5ff4dbaa8e1e8ad8fb33e82 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76202b0d43a541dba0f0298baf02341d : SUCCESS in 5m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e4b84db3c2aa4763adc1f1b8cbc643c8 : SUCCESS in 1h 20m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bdc2dc3fbb964fe5895bfe0de2dae0a7 : SUCCESS in 4m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac1bc4a9983048349ed26bf3e3b6b09b : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":1},{"id":"8ef30d4df5ff1e1e8b4d98c6b962bf7fd583856a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 21:52:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8a327046d5a80aaadd285870f78ff47200f9a701","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 21:54:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb19390cde604884b269fccf0b4e71b3 : SUCCESS in 1m 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eaf4ccdf05684929b8d8406187de5c28 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7900aa42a22ea8962fb33af680f636c20bf39119","revisions":{"7900aa42a22ea8962fb33af680f636c20bf39119":{"kind":"REWORK","_number":1,"created":"2019-04-25 09:49:48.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/48/655648/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/48/655648/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/48/655648/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/48/655648/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/48/655648/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/48/655648/1"}}},"commit":{"parents":[{"commit":"f029a1808f02ded25ea08af1d81704a0c1b7cad3","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/f029a1808f02ded25ea08af1d81704a0c1b7cad3"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2019-04-25 09:39:36.000000000","tz":180},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2019-04-25 09:49:42.000000000","tz":180},"subject":"Use upper constraints from opendev.org","message":"Use upper constraints from opendev.org\n\noslo.cache uses a simple curl command in its tox_install.sh\nto fetch the upper-constraints, and curl does not follow redirects\nthat are now in place due to move to opendev.org.\n\nThis breaks running tox-based tests w/o overriding the\nUPPER_CONSTRAINTS_FILE which upstream CI does,\nfor example running tox locally.\n\nChange-Id: Ifcb555c8f4ccb813b48c613c79dceb4bf23318eb\n(cherry picked from commit 6eceb160eaaebbc6ea04768f62876738aa61c2d1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/7900aa42a22ea8962fb33af680f636c20bf39119"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/7900aa42a22ea8962fb33af680f636c20bf39119"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
