)]}'
{"id":"openstack%2Foslo.log~655649","triplet_id":"openstack%2Foslo.log~stable%2Fqueens~Ifcb555c8f4ccb813b48c613c79dceb4bf23318eb","project":"openstack/oslo.log","branch":"stable/queens","topic":"opendev-u-c","hashtags":[],"change_id":"Ifcb555c8f4ccb813b48c613c79dceb4bf23318eb","subject":"Use upper constraints from opendev.org","status":"MERGED","created":"2019-04-25 09:50:58.000000000","updated":"2019-10-02 20:19:30.000000000","submitted":"2019-10-02 20:18:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655649-1570047493887-fdea6b69","meta_rev_id":"5e7ffd778563d5d969e407f9df50e168f4082596","_number":655649,"virtual_id_number":655649,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"date":"2019-10-02 17:22:03.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-02 20:18:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":2,"date":"2019-05-20 07:47:34.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2019-05-03 16:38:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-05-03 00:00:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":1,"date":"2019-05-20 07:47:34.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-03 00:00:07.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-05-03 16:38:45.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-20 07:47:34.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2019-10-02 17:22:03.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2019-10-02 20:18:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ef342744842be5d1316046ce07bb54b0fb748614","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2019-04-25 09:50:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13f1d98a51151d32dff4db8f0f119b65509a5588","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 11:52:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/49/655649/1/check/openstack-tox-pep8/32841ee/ : SUCCESS in 4m 24s\n- openstack-tox-py27 http://logs.openstack.org/49/655649/1/check/openstack-tox-py27/a77f998/ : SUCCESS in 5m 02s\n- openstack-tox-py35 http://logs.openstack.org/49/655649/1/check/openstack-tox-py35/764629a/ : SUCCESS in 5m 13s\n- openstack-tox-docs http://logs.openstack.org/49/655649/1/check/openstack-tox-docs/92cf7bc/html/ : SUCCESS in 4m 53s\n- tempest-full http://logs.openstack.org/49/655649/1/check/tempest-full/15dc517/ : SUCCESS in 1h 28m 45s\n- build-openstack-releasenotes http://logs.openstack.org/49/655649/1/check/build-openstack-releasenotes/5386e9a/html/ : SUCCESS in 4m 41s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/49/655649/1/check/oslo.log-src-grenade-devstack/2b1af38/ : SUCCESS in 53m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4cdea27e64dca7a7625b958e01c665118ecbc2e0","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-05-03 00:00:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cf2c96e9f5e47eeb5fcb8827a9a9cd8dbed532a1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 16:38:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e1e0e4ce06924e758f4a6677bb9438423cc834d9","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2019-05-20 07:47:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9d9437d73489a2a8f54fe9f73f5ef3a5f9367cd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 07:47:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed17131f8dc2a0c659dead64fbf63f2587716a0f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 09:43:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/49/655649/1/gate/openstack-tox-pep8/0d79870/ : FAILURE in 5m 29s\n- openstack-tox-py27 http://logs.openstack.org/49/655649/1/gate/openstack-tox-py27/80bcca9/ : SUCCESS in 4m 29s\n- openstack-tox-py35 http://logs.openstack.org/49/655649/1/gate/openstack-tox-py35/bcadbb6/ : SUCCESS in 5m 55s\n- openstack-tox-docs http://logs.openstack.org/49/655649/1/gate/openstack-tox-docs/da76a8c/html/ : SUCCESS in 4m 20s\n- tempest-full http://logs.openstack.org/49/655649/1/gate/tempest-full/4e22c1e/ : SUCCESS in 1h 22m 17s\n- build-openstack-releasenotes http://logs.openstack.org/49/655649/1/gate/build-openstack-releasenotes/604de9b/html/ : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"f232235177558a0df3da2a96fc1f06c5e1875233","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-10-02 17:22:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3508c4d918e13cab69040b489dc6e46ce30a92ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 18:46:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1ad4992f3d64f8796c0b89e4aa82234 : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1c7355405a26400e9618a24629cc4795 : SUCCESS in 4m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b3e15ee7fc614bcbb5ffe553d7eeeed2 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ce06349fe374c6285391cd1b1fa899a : SUCCESS in 3m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3dcb0fbb05db447eba07daa9a1f15429 : SUCCESS in 1h 19m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2458773375e4d5a84613548d27deda0 : SUCCESS in 3m 27s\n- oslo.log-src-grenade-devstack https://zuul.opendev.org/t/openstack/build/de804b91629d47eab3b5ab31d66e46c1 : SUCCESS in 1h 02m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf4f372553d7d4ef91d3efb68aaf8bf741a55afb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 18:47:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e36901b7cd40e774e51f80d8e34baf93eac0e15b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 20:18:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8124c7557e8e426abb74245ef9ef6951 : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a7f02ad39fb34976a1ac7c2128eaa671 : SUCCESS in 3m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e7ac8ef7da1c4850bc0badea10540c9b : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e19347e77fb24ce8aa11f95d1c9910b4 : SUCCESS in 3m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7a22e544954244548434d0ec2024c11b : SUCCESS in 1h 25m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eea6628d6eb349c2b29705b1bb8aac9c : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"22f68edf9e3e82dfe1c38e99ad35901ec1ca2f75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 20:18:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a69b78f2057409cdf4cd85a0b586f4e6f10b6e4e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 20:19:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7352d40341c24731a793c3c06cbb557b : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0beb820761d425f836fbd688b9dfc9b : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4a655241e87fc908b36fab8fa480e33406bed7bd","revisions":{"4a655241e87fc908b36fab8fa480e33406bed7bd":{"kind":"REWORK","_number":1,"created":"2019-04-25 09:50:58.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/49/655649/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/49/655649/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/49/655649/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/49/655649/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/49/655649/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/49/655649/1"}}},"commit":{"parents":[{"commit":"afebcd03f77bf2853fcf50ccb8dbf6f5ab5a6161","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/afebcd03f77bf2853fcf50ccb8dbf6f5ab5a6161"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2019-04-25 09:39:36.000000000","tz":180},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2019-04-25 09:50:46.000000000","tz":180},"subject":"Use upper constraints from opendev.org","message":"Use upper constraints from opendev.org\n\noslo.cache uses a simple curl command in its tox_install.sh\nto fetch the upper-constraints, and curl does not follow redirects\nthat are now in place due to move to opendev.org.\n\nThis breaks running tox-based tests w/o overriding the\nUPPER_CONSTRAINTS_FILE which upstream CI does,\nfor example running tox locally.\n\nChange-Id: Ifcb555c8f4ccb813b48c613c79dceb4bf23318eb\n(cherry picked from commit 6eceb160eaaebbc6ea04768f62876738aa61c2d1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/4a655241e87fc908b36fab8fa480e33406bed7bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/4a655241e87fc908b36fab8fa480e33406bed7bd"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
