)]}'
{"id":"openstack%2Foslo.log~657040","triplet_id":"openstack%2Foslo.log~master~I67c50430ced6efcc2f29ec7d8adc871415922d6f","project":"openstack/oslo.log","branch":"master","topic":"cap_bandit_only","hashtags":[],"change_id":"I67c50430ced6efcc2f29ec7d8adc871415922d6f","subject":"Avoid tox_install.sh for constraints support","status":"MERGED","created":"2019-05-03 20:50:41.000000000","updated":"2019-05-15 17:44:04.000000000","submitted":"2019-05-15 17:44:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657040-1557942244506-cf8f5779","meta_rev_id":"1a3ba9a1abdc1ee3290af2f2718a78dd765b7465","_number":657040,"virtual_id_number":657040,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"date":"2019-05-14 16:52:51.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-05-15 17:44:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-05-14 18:42:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-05-15 15:40:31.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":1,"date":"2019-05-14 17:29:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-05-15 15:40:31.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-14 09:49:58.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-05-14 14:59:02.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-05-14 16:52:51.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-05-14 17:29:15.000000000","updated_by":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"reviewer":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"state":"REVIEWER"},{"updated":"2019-05-14 18:42:45.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-05-15 15:40:31.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2019-05-15 17:44:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"af32aa4fa15cc2c24e988089ec568065eebd6316","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 20:50:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ec41f51a0b5ce7a7722acba2b557cd27f38e569","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 22:53:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/40/657040/1/check/requirements-check/5f460ce/ : FAILURE in 3m 20s\n- tempest-full http://logs.openstack.org/40/657040/1/check/tempest-full/e83af36/ : FAILURE in 1h 57m 44s\n- tempest-full-py3 http://logs.openstack.org/40/657040/1/check/tempest-full-py3/89457ee/ : SUCCESS in 1h 24m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/657040/1/check/openstack-tox-lower-constraints/decaf99/ : SUCCESS in 6m 05s\n- openstack-tox-pep8 http://logs.openstack.org/40/657040/1/check/openstack-tox-pep8/9479ae8/ : SUCCESS in 7m 08s\n- openstack-tox-py27 http://logs.openstack.org/40/657040/1/check/openstack-tox-py27/21f5f3b/ : SUCCESS in 4m 01s\n- openstack-tox-py36 http://logs.openstack.org/40/657040/1/check/openstack-tox-py36/bbfa569/ : SUCCESS in 4m 43s\n- openstack-tox-docs http://logs.openstack.org/40/657040/1/check/openstack-tox-docs/73faa4e/html/ : SUCCESS in 4m 42s\n- build-openstack-releasenotes http://logs.openstack.org/40/657040/1/check/build-openstack-releasenotes/2fdf868/html/ : SUCCESS in 7m 01s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/40/657040/1/check/oslo.log-src-grenade-devstack/ed72f1d/ : SUCCESS in 1h 02m 48s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/40/657040/1/check/oslo.log-jsonformatter/5adcc33/ : SUCCESS in 52m 51s","accounts_in_message":[],"_revision_number":1},{"id":"47b9589b91b4d69415817626fb9905a770658215","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-09 10:33:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"51adc2bad2d7033ec1196e48d86c99c4c8d24ded","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 12:11:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/40/657040/2/check/requirements-check/9e0e19a/ : FAILURE in 3m 22s\n- tempest-full http://logs.openstack.org/40/657040/2/check/tempest-full/f221e27/ : SUCCESS in 1h 30m 13s\n- tempest-full-py3 http://logs.openstack.org/40/657040/2/check/tempest-full-py3/041417a/ : SUCCESS in 1h 31m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/657040/2/check/openstack-tox-lower-constraints/1093737/ : SUCCESS in 6m 58s\n- openstack-tox-pep8 http://logs.openstack.org/40/657040/2/check/openstack-tox-pep8/b1a009f/ : FAILURE in 4m 15s\n- openstack-tox-py27 http://logs.openstack.org/40/657040/2/check/openstack-tox-py27/5728f2a/ : SUCCESS in 4m 14s\n- openstack-tox-py36 http://logs.openstack.org/40/657040/2/check/openstack-tox-py36/80d83e7/ : SUCCESS in 4m 06s\n- openstack-tox-docs http://logs.openstack.org/40/657040/2/check/openstack-tox-docs/5e056d8/html/ : SUCCESS in 4m 09s\n- build-openstack-releasenotes http://logs.openstack.org/40/657040/2/check/build-openstack-releasenotes/35cc16a/html/ : SUCCESS in 4m 11s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/40/657040/2/check/oslo.log-src-grenade-devstack/f7ec655/ : SUCCESS in 59m 56s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/40/657040/2/check/oslo.log-jsonformatter/b29f9d3/ : SUCCESS in 46m 37s","accounts_in_message":[],"_revision_number":2},{"id":"ab88d5be43278ef4545a6d33c0b150560a1f270b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-09 13:33:16.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"6bd5abf26f1530c0260e4e0827f9e0d25fedaca6","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-09 13:45:33.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b463f2feb05e0ff6eb67dd4febb40b54183a3555","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 15:23:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/40/657040/3/check/requirements-check/011ddee/ : SUCCESS in 2m 56s\n- tempest-full http://logs.openstack.org/40/657040/3/check/tempest-full/c59240d/ : SUCCESS in 1h 43m 46s\n- tempest-full-py3 http://logs.openstack.org/40/657040/3/check/tempest-full-py3/ee9682a/ : SUCCESS in 1h 35m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/657040/3/check/openstack-tox-lower-constraints/e335c84/ : SUCCESS in 5m 08s\n- openstack-tox-pep8 http://logs.openstack.org/40/657040/3/check/openstack-tox-pep8/1de2807/ : FAILURE in 3m 43s\n- openstack-tox-py27 http://logs.openstack.org/40/657040/3/check/openstack-tox-py27/810b0e4/ : SUCCESS in 8m 57s\n- openstack-tox-py36 http://logs.openstack.org/40/657040/3/check/openstack-tox-py36/6d0d8fc/ : SUCCESS in 9m 52s\n- openstack-tox-docs http://logs.openstack.org/40/657040/3/check/openstack-tox-docs/dc68675/html/ : SUCCESS in 5m 13s\n- build-openstack-releasenotes http://logs.openstack.org/40/657040/3/check/build-openstack-releasenotes/929c28a/html/ : SUCCESS in 4m 25s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/40/657040/3/check/oslo.log-src-grenade-devstack/badfd00/ : SUCCESS in 1h 03m 35s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/40/657040/3/check/oslo.log-jsonformatter/6422678/ : SUCCESS in 44m 35s","accounts_in_message":[],"_revision_number":3},{"id":"2b3ec0ea69f9162284ce3cfc9f89ae3ae1143923","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-10 09:53:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c7497114452387ea551373a3f4dc2232d8d417a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 12:03:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/40/657040/4/check/requirements-check/038353f/ : SUCCESS in 5m 00s\n- tempest-full http://logs.openstack.org/40/657040/4/check/tempest-full/f45814a/ : SUCCESS in 1h 53m 30s\n- tempest-full-py3 http://logs.openstack.org/40/657040/4/check/tempest-full-py3/accc9a3/ : SUCCESS in 1h 31m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/657040/4/check/openstack-tox-lower-constraints/ecfa48d/ : SUCCESS in 6m 11s\n- openstack-tox-pep8 http://logs.openstack.org/40/657040/4/check/openstack-tox-pep8/5a5dcf0/ : SUCCESS in 4m 21s\n- openstack-tox-py27 http://logs.openstack.org/40/657040/4/check/openstack-tox-py27/e32270a/ : SUCCESS in 6m 01s\n- openstack-tox-py36 http://logs.openstack.org/40/657040/4/check/openstack-tox-py36/6a002e3/ : SUCCESS in 4m 46s\n- openstack-tox-docs http://logs.openstack.org/40/657040/4/check/openstack-tox-docs/7bd4877/html/ : SUCCESS in 5m 47s\n- build-openstack-releasenotes http://logs.openstack.org/40/657040/4/check/build-openstack-releasenotes/aa60a05/html/ : SUCCESS in 5m 34s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/40/657040/4/check/oslo.log-src-grenade-devstack/1643105/ : SUCCESS in 58m 26s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/40/657040/4/check/oslo.log-jsonformatter/d36ec26/ : SUCCESS in 44m 23s","accounts_in_message":[],"_revision_number":4},{"id":"c273edcb8e965d6c476e7719098af0a9244ce888","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-05-13 14:35:10.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"4d0f022e3a40e47a89ef27084ec28bed73c4cdd8","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-05-14 09:49:58.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"d11ad98740fb31e5dcd63ff97a8cd19916ce9940","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-05-14 12:24:38.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"51667d52b2a275c998d93051b6a05c929c774484","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-05-14 14:59:02.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"28150f8c366d373f20075296e53fe02b55bcefd6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-14 16:33:43.000000000","message":"Patch Set 4:\n\nNeed to base rebased over https://review.opendev.org/659055 since\n- https://review.opendev.org/#/c/658079/2 (fix monotonic)\n- https://review.opendev.org/#/c/658319/1 (fix bandit)\nneeds to be fixed together by https://review.opendev.org/659055","accounts_in_message":[],"_revision_number":4},{"id":"b85ed7d7206843247c9a10420a6c9ed0ac0bed27","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-14 16:48:20.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"16369b02bbce64b065d778830a8e35232e0fcb03","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-14 16:52:51.000000000","message":"Patch Set 5:\n\nRebased on \u003d\u003e https://review.opendev.org/#/c/659055/6 to fix the CI requirements check all in one and avoid cross patches CI errors","accounts_in_message":[],"_revision_number":5},{"id":"1cc7198e114a187ac00e9324caf3bd6f5e9a7358","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-05-14 17:29:15.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"5aac6a91c928fe344af669c5768c1b96f9c5d040","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 18:31:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/40/657040/5/check/requirements-check/cdba29b/ : SUCCESS in 4m 26s\n- tempest-full http://logs.openstack.org/40/657040/5/check/tempest-full/d900963/ : SUCCESS in 1h 34m 57s\n- tempest-full-py3 http://logs.openstack.org/40/657040/5/check/tempest-full-py3/81a75d3/ : SUCCESS in 1h 41m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/657040/5/check/openstack-tox-lower-constraints/c847b6c/ : SUCCESS in 4m 17s\n- openstack-tox-pep8 http://logs.openstack.org/40/657040/5/check/openstack-tox-pep8/0a3d3e3/ : SUCCESS in 4m 21s\n- openstack-tox-py27 http://logs.openstack.org/40/657040/5/check/openstack-tox-py27/5a06778/ : SUCCESS in 4m 50s\n- openstack-tox-py36 http://logs.openstack.org/40/657040/5/check/openstack-tox-py36/3c71c20/ : SUCCESS in 5m 16s\n- openstack-tox-docs http://logs.openstack.org/40/657040/5/check/openstack-tox-docs/8a121d9/html/ : SUCCESS in 5m 04s\n- build-openstack-releasenotes http://logs.openstack.org/40/657040/5/check/build-openstack-releasenotes/10f3371/html/ : SUCCESS in 3m 57s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/40/657040/5/check/oslo.log-src-grenade-devstack/1e7905f/ : SUCCESS in 58m 09s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/40/657040/5/check/oslo.log-jsonformatter/3d60302/ : SUCCESS in 44m 05s","accounts_in_message":[],"_revision_number":5},{"id":"b47f3271ee84cd3185adb48d0b446cde1d5af8bf","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-14 18:42:45.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"58ae91acad416726e3338f7fe1c67ee311689e36","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-05-15 15:40:31.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"eeeace4633198f209a77c4ba1ab1da0beeb0e7c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 15:40:43.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e0476eb0a3f1a59f9fa36672ce0aa947e7440eff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 17:44:04.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/40/657040/5/gate/requirements-check/7f0bbab/ : SUCCESS in 4m 13s\n- tempest-full http://logs.openstack.org/40/657040/5/gate/tempest-full/7b098c4/ : SUCCESS in 1h 56m 10s\n- tempest-full-py3 http://logs.openstack.org/40/657040/5/gate/tempest-full-py3/838f365/ : SUCCESS in 1h 40m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/40/657040/5/gate/openstack-tox-lower-constraints/b19870a/ : SUCCESS in 4m 09s\n- openstack-tox-pep8 http://logs.openstack.org/40/657040/5/gate/openstack-tox-pep8/477ae90/ : SUCCESS in 4m 36s\n- openstack-tox-py27 http://logs.openstack.org/40/657040/5/gate/openstack-tox-py27/2f7055b/ : SUCCESS in 4m 31s\n- openstack-tox-py36 http://logs.openstack.org/40/657040/5/gate/openstack-tox-py36/266db55/ : SUCCESS in 5m 28s\n- openstack-tox-docs http://logs.openstack.org/40/657040/5/gate/openstack-tox-docs/12e7985/html/ : SUCCESS in 4m 43s\n- build-openstack-releasenotes http://logs.openstack.org/40/657040/5/gate/build-openstack-releasenotes/bc9eb89/html/ : SUCCESS in 3m 59s\n- oslo.log-jsonformatter http://logs.openstack.org/40/657040/5/gate/oslo.log-jsonformatter/fd71055/ : SUCCESS in 40m 20s","accounts_in_message":[],"_revision_number":5},{"id":"f30d3321da7b1ec477c1efeedbb70bfee475d108","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 17:44:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"8c5317152b0811ef11d3078b67e30b36c83bb6e8","revisions":{"16007bb4f2d8d39d0196d02d9b49e05a6c4f8f1f":{"kind":"REWORK","_number":1,"created":"2019-05-03 20:50:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/657040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/40/657040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/40/657040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/40/657040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/40/657040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/40/657040/1"}}},"commit":{"parents":[{"commit":"d71bbcb04edb6b3f268abea639641dcf4228d0a0","subject":"Merge \"Add TRAIN to deprecated releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/d71bbcb04edb6b3f268abea639641dcf4228d0a0"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-03 20:47:10.000000000","tz":-360},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-03 20:49:58.000000000","tz":-360},"subject":"Avoid tox_install.sh for constraints support","message":"Avoid tox_install.sh for constraints support\n\nWe do not need tox_install.sh, pip can handle constraints itself\nand install the project correctly. Thus update tox.ini and remove\nthe now obsolete tools/tox_install.sh file.\n\nThis follows https://review.openstack.org/#/c/508061 to remove\ntools/tox_install.sh.\n\nChange-Id: I67c50430ced6efcc2f29ec7d8adc871415922d6f\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/16007bb4f2d8d39d0196d02d9b49e05a6c4f8f1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/16007bb4f2d8d39d0196d02d9b49e05a6c4f8f1f"}]},"branch":"refs/heads/master"},"f79a834600dce39f4714130672fb5461b708757f":{"kind":"REWORK","_number":2,"created":"2019-05-09 10:33:21.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/657040/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/40/657040/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/40/657040/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/40/657040/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/40/657040/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/40/657040/2"}}},"commit":{"parents":[{"commit":"29671ef2bfacb416d397abc57170bb090b116f68","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/29671ef2bfacb416d397abc57170bb090b116f68"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-03 20:47:10.000000000","tz":-360},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-09 10:32:33.000000000","tz":60},"subject":"Avoid tox_install.sh for constraints support","message":"Avoid tox_install.sh for constraints support\n\nWe do not need tox_install.sh, pip can handle constraints itself\nand install the project correctly. Thus update tox.ini and remove\nthe now obsolete tools/tox_install.sh file.\n\nThis follows https://review.openstack.org/#/c/508061 to remove\ntools/tox_install.sh.\n\nChange-Id: I67c50430ced6efcc2f29ec7d8adc871415922d6f\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/f79a834600dce39f4714130672fb5461b708757f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/f79a834600dce39f4714130672fb5461b708757f"}]},"branch":"refs/heads/master"},"03a7b52874927a830fcfae23000729bf872b944a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-05-09 13:33:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/657040/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/40/657040/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/40/657040/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/40/657040/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/40/657040/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/40/657040/3"}}},"commit":{"parents":[{"commit":"7d6878a6b9416fa1d728c88f38d5ba4b3d94bf5a","subject":"Limit monotonic to python \u003c 3.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/7d6878a6b9416fa1d728c88f38d5ba4b3d94bf5a"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-03 20:47:10.000000000","tz":-360},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-09 13:33:04.000000000","tz":60},"subject":"Avoid tox_install.sh for constraints support","message":"Avoid tox_install.sh for constraints support\n\nWe do not need tox_install.sh, pip can handle constraints itself\nand install the project correctly. Thus update tox.ini and remove\nthe now obsolete tools/tox_install.sh file.\n\nThis follows https://review.openstack.org/#/c/508061 to remove\ntools/tox_install.sh.\n\nChange-Id: I67c50430ced6efcc2f29ec7d8adc871415922d6f\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/03a7b52874927a830fcfae23000729bf872b944a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/03a7b52874927a830fcfae23000729bf872b944a"}]},"branch":"refs/heads/master"},"ba56ef07ad90b729c6f4fdadb1668feae7d89abe":{"kind":"REWORK","_number":4,"created":"2019-05-10 09:53:55.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/657040/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/40/657040/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/40/657040/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/40/657040/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/40/657040/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/40/657040/4"}}},"commit":{"parents":[{"commit":"bd8f89cbd60e646d8b3134ecf59f159e4a751feb","subject":"Limit monotonic to python \u003c 3.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/bd8f89cbd60e646d8b3134ecf59f159e4a751feb"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-03 20:47:10.000000000","tz":-360},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-10 09:53:49.000000000","tz":60},"subject":"Avoid tox_install.sh for constraints support","message":"Avoid tox_install.sh for constraints support\n\nWe do not need tox_install.sh, pip can handle constraints itself\nand install the project correctly. Thus update tox.ini and remove\nthe now obsolete tools/tox_install.sh file.\n\nThis follows https://review.openstack.org/#/c/508061 to remove\ntools/tox_install.sh.\n\nChange-Id: I67c50430ced6efcc2f29ec7d8adc871415922d6f\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/ba56ef07ad90b729c6f4fdadb1668feae7d89abe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/ba56ef07ad90b729c6f4fdadb1668feae7d89abe"}]},"branch":"refs/heads/master"},"8c5317152b0811ef11d3078b67e30b36c83bb6e8":{"kind":"REWORK","_number":5,"created":"2019-05-14 16:48:20.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/40/657040/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/40/657040/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/40/657040/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/40/657040/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/40/657040/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/40/657040/5"}}},"commit":{"parents":[{"commit":"1d4c0f42f941294612b9548058dbe39d21047bba","subject":"Cap bandit below 1.6.0 version and update sphinx and limit monotonic.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1d4c0f42f941294612b9548058dbe39d21047bba"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-03 20:47:10.000000000","tz":-360},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-14 16:47:48.000000000","tz":120},"subject":"Avoid tox_install.sh for constraints support","message":"Avoid tox_install.sh for constraints support\n\nWe do not need tox_install.sh, pip can handle constraints itself\nand install the project correctly. Thus update tox.ini and remove\nthe now obsolete tools/tox_install.sh file.\n\nThis follows https://review.openstack.org/#/c/508061 to remove\ntools/tox_install.sh.\n\nChange-Id: I67c50430ced6efcc2f29ec7d8adc871415922d6f\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/8c5317152b0811ef11d3078b67e30b36c83bb6e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/8c5317152b0811ef11d3078b67e30b36c83bb6e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
