)]}'
{"id":"openstack%2Foslo.log~658354","triplet_id":"openstack%2Foslo.log~master~I22099ec81bc13d8056385722d21a22e21cad448f","project":"openstack/oslo.log","branch":"master","topic":"fix_bandit_command","hashtags":[],"change_id":"I22099ec81bc13d8056385722d21a22e21cad448f","subject":"Fix bandit command.","status":"ABANDONED","created":"2019-05-10 13:00:21.000000000","updated":"2019-05-13 09:37:19.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6ee3f9dfd27ac758efa531cecc5560f14f913f5a","_number":658354,"virtual_id_number":658354,"owner":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2019-05-10 17:29:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":-1,"date":"2019-05-13 09:18:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-10 17:29:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-13 09:18:49.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"53af3263e588a16b92a21e98c17f37919ce83057","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-05-10 13:00:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8ad5dbccc9d450d84a04fa680f1787e9e79ab6d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-10 15:09:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e614e7f2199bf4482cafe2221afb6e65d18c4f89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 15:10:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/54/658354/1/check/tempest-full/3fb38b8/ : SUCCESS in 2h 01m 31s\n- tempest-full-py3 http://logs.openstack.org/54/658354/1/check/tempest-full-py3/9463fc6/ : SUCCESS in 1h 32m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/54/658354/1/check/openstack-tox-lower-constraints/f047b9a/ : SUCCESS in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/54/658354/1/check/openstack-tox-pep8/510b559/ : SUCCESS in 4m 35s\n- openstack-tox-py27 http://logs.openstack.org/54/658354/1/check/openstack-tox-py27/4342138/ : SUCCESS in 4m 57s\n- openstack-tox-py36 http://logs.openstack.org/54/658354/1/check/openstack-tox-py36/de0129e/ : SUCCESS in 5m 00s\n- openstack-tox-docs http://logs.openstack.org/54/658354/1/check/openstack-tox-docs/e6d3acb/html/ : SUCCESS in 4m 49s\n- build-openstack-releasenotes http://logs.openstack.org/54/658354/1/check/build-openstack-releasenotes/8f7af95/html/ : SUCCESS in 4m 47s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/54/658354/1/check/oslo.log-src-grenade-devstack/b7ce6cf/ : SUCCESS in 58m 12s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/54/658354/1/check/oslo.log-jsonformatter/4c61961/ : SUCCESS in 38m 13s","accounts_in_message":[],"_revision_number":1},{"id":"676fbaa0d321e0d00febd705a52ef03106867c6e","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-05-10 15:22:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3279089ce20e4310404aeca3205f2569e81efe3f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-10 15:31:49.000000000","message":"Patch Set 2: Code-Review+2\n\nTested locally work for me","accounts_in_message":[],"_revision_number":2},{"id":"fa30ac119566dca55c2651f0a6ccc9d69cb657e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 17:29:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/54/658354/2/check/tempest-full/81857e3/ : SUCCESS in 2h 05m 40s\n- tempest-full-py3 http://logs.openstack.org/54/658354/2/check/tempest-full-py3/39d49a8/ : SUCCESS in 1h 30m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/54/658354/2/check/openstack-tox-lower-constraints/e711c35/ : SUCCESS in 4m 16s\n- openstack-tox-pep8 http://logs.openstack.org/54/658354/2/check/openstack-tox-pep8/2f58a2e/ : SUCCESS in 4m 13s\n- openstack-tox-py27 http://logs.openstack.org/54/658354/2/check/openstack-tox-py27/2d78fdc/ : SUCCESS in 4m 47s\n- openstack-tox-py36 http://logs.openstack.org/54/658354/2/check/openstack-tox-py36/acdbf2c/ : SUCCESS in 4m 56s\n- openstack-tox-docs http://logs.openstack.org/54/658354/2/check/openstack-tox-docs/a97b624/html/ : SUCCESS in 4m 36s\n- build-openstack-releasenotes http://logs.openstack.org/54/658354/2/check/build-openstack-releasenotes/2247bd3/html/ : SUCCESS in 4m 20s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/54/658354/2/check/oslo.log-src-grenade-devstack/f1056c9/ : SUCCESS in 1h 10m 29s (non-voting)\n- oslo.log-jsonformatter http://logs.openstack.org/54/658354/2/check/oslo.log-jsonformatter/79af673/ : SUCCESS in 42m 15s","accounts_in_message":[],"_revision_number":2},{"id":"d678104b86b65865dcc4316b37c79b4a003eadad","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-13 09:05:29.000000000","message":"Patch Set 2:\n\nWarning! Duplicated with \u003d\u003e https://review.opendev.org/#/c/658319/1","accounts_in_message":[],"_revision_number":2},{"id":"9becdbac27f232827260e06a764123aa89338e14","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-13 09:18:49.000000000","message":"Patch Set 2: Code-Review-1\n\nBandit version need to be bumped too.\n\nI suggest to you to try to collaborate with Stephen on this other patch (https://review.opendev.org/#/c/658319/1) to fix the issue.\n\nAlso to fix the stephen CI patch I guess we need to bump bandit on openstack/requirements [1] too. \n\n[1] https://github.com/openstack/requirements/blob/6bf8756f6102fa4538501d3344ed915d29f3af67/openstack_requirements/tests/files/upper-constraints.txt#L21","accounts_in_message":[],"_revision_number":2},{"id":"181ee9a0aa5ba79822ef0d2551e7445749cb1d90","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-05-13 09:37:19.000000000","message":"Abandoned\n\nAlready proposed.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b2c5e299a0f73933f08d5180682d904a6d49206e","revisions":{"eb9fc9d7a983d08e64c7d4ffd8b22ce4bdb482ff":{"kind":"REWORK","_number":1,"created":"2019-05-10 13:00:21.000000000","uploader":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"ref":"refs/changes/54/658354/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/54/658354/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/54/658354/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/54/658354/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/54/658354/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/54/658354/1"}}},"commit":{"parents":[{"commit":"29671ef2bfacb416d397abc57170bb090b116f68","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/29671ef2bfacb416d397abc57170bb090b116f68"}]}],"author":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-05-10 12:59:28.000000000","tz":120},"committer":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-05-10 12:59:28.000000000","tz":120},"subject":"Fix bandit command.","message":"Fix bandit command.\n\nFix the path of test directory to bandit command.\n\nChange-Id: I22099ec81bc13d8056385722d21a22e21cad448f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/eb9fc9d7a983d08e64c7d4ffd8b22ce4bdb482ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/eb9fc9d7a983d08e64c7d4ffd8b22ce4bdb482ff"}]},"branch":"refs/heads/master"},"b2c5e299a0f73933f08d5180682d904a6d49206e":{"kind":"REWORK","_number":2,"created":"2019-05-10 15:22:00.000000000","uploader":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"ref":"refs/changes/54/658354/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/54/658354/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/54/658354/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/54/658354/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/54/658354/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/54/658354/2"}}},"commit":{"parents":[{"commit":"29671ef2bfacb416d397abc57170bb090b116f68","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/29671ef2bfacb416d397abc57170bb090b116f68"}]}],"author":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-05-10 12:59:28.000000000","tz":120},"committer":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-05-10 15:21:47.000000000","tz":120},"subject":"Fix bandit command.","message":"Fix bandit command.\n\nFix the path of test directory to bandit command.\n\nChange-Id: I22099ec81bc13d8056385722d21a22e21cad448f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/b2c5e299a0f73933f08d5180682d904a6d49206e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/b2c5e299a0f73933f08d5180682d904a6d49206e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
