)]}'
{"id":"openstack%2Foslo.log~659056","triplet_id":"openstack%2Foslo.log~master~I14c4db063c0f4f65279cb737ffbf5212f572ad81","project":"openstack/oslo.log","branch":"master","topic":"fix_sphinx_version","hashtags":[],"change_id":"I14c4db063c0f4f65279cb737ffbf5212f572ad81","subject":"Update sphinx requirement.","status":"ABANDONED","created":"2019-05-14 12:38:15.000000000","updated":"2019-05-14 15:12:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f94f82c45e01cdc5ae2ec6ebe9aaf6eb4e528fdd","_number":659056,"virtual_id_number":659056,"owner":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":-1,"date":"2019-05-14 13:40:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":1,"date":"2019-05-14 12:56:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":-1,"date":"2019-05-14 14:56:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-14 12:56:33.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2019-05-14 13:40:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-14 14:56:25.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"de432558821467926be2e19fdd6f94abccea1ed4","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-05-14 12:38:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14271d07355d9bf85fc4ba08f886fd343a9e352e","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-05-14 12:56:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"02ab7c3c06c401f256b89bbfa92e877b4d18e73f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 13:40:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/56/659056/1/check/requirements-check/f0516dd/ : FAILURE in 3m 17s\n- openstack-tox-pep8 http://logs.openstack.org/56/659056/1/check/openstack-tox-pep8/d33c799/ : FAILURE in 4m 34s\n- openstack-tox-docs http://logs.openstack.org/56/659056/1/check/openstack-tox-docs/fc2b6b1/ : POST_FAILURE in 4m 25s\n- build-openstack-releasenotes http://logs.openstack.org/56/659056/1/check/build-openstack-releasenotes/ed086af/ : RETRY_LIMIT in 3m 43s\n- oslo.log-src-grenade-devstack http://logs.openstack.org/56/659056/1/check/oslo.log-src-grenade-devstack/2adb865/ : SUCCESS in 56m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1920fc24def68822b95b14a1acd9d28c3150f708","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-14 14:56:25.000000000","message":"Patch Set 1: Code-Review-1\n\nas I sayed to you on IRC you need to merge this one with https://review.opendev.org/659055 ...\n\nThe both need to be fixed together to avoid cross error due to sphinx and bandit.\n\nhttp://logs.openstack.org/56/659056/1/check/openstack-tox-pep8/d33c799/job-output.txt.gz#_2019-05-14_12_43_02_763432","accounts_in_message":[],"_revision_number":1},{"id":"75e43c38f48a5146b68a1c38a770a7a37515c786","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-14 15:12:32.000000000","message":"Abandoned\n\nAbandoned see: https://review.opendev.org/#/c/659055/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bf924c524af73232fc2389bdef0589ee6355f88d","revisions":{"bf924c524af73232fc2389bdef0589ee6355f88d":{"kind":"REWORK","_number":1,"created":"2019-05-14 12:38:15.000000000","uploader":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"ref":"refs/changes/56/659056/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/56/659056/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/56/659056/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/56/659056/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/56/659056/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/56/659056/1"}}},"commit":{"parents":[{"commit":"29671ef2bfacb416d397abc57170bb090b116f68","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/29671ef2bfacb416d397abc57170bb090b116f68"}]}],"author":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-05-14 12:38:07.000000000","tz":120},"committer":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-05-14 12:38:07.000000000","tz":120},"subject":"Update sphinx requirement.","message":"Update sphinx requirement.\n\nSphinx 2.0 no longer works on python 2.7, so need to start capping it\nthere as well.\n\nChange-Id: I14c4db063c0f4f65279cb737ffbf5212f572ad81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/bf924c524af73232fc2389bdef0589ee6355f88d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/bf924c524af73232fc2389bdef0589ee6355f88d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
